)]}'
{"id":"openstack%2Freleases~652877","triplet_id":"openstack%2Freleases~master~Iea97c3534df7d73d2666c4765dfc50eb4c99f306","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"Iea97c3534df7d73d2666c4765dfc50eb4c99f306","subject":"[congress] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:57:56.000000000","updated":"2019-04-22 19:05:04.000000000","submitted":"2019-04-22 19:05:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652877-1555959904865-61dee02a","meta_rev_id":"a5e0d9ac5ed3e7370a6a540f28763fb8295f7c30","_number":652877,"virtual_id_number":652877,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-04-22 19:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-22 18:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-19 00:36:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-04-22 18:52:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 18:01:57.000000000","updated_by":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"reviewer":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"state":"REVIEWER"},{"updated":"2019-04-22 18:52:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-22 19:05:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7f5df4ee36dab7bb42f0fa1648637d3faba5a06","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fa3a43c6678b4d5b773418abea5274e93b7f3cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 07:26:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/77/652877/1/check/openstack-tox-docs/3335e1d/html/ : SUCCESS in 7m 48s\n- openstack-tox-validate http://logs.openstack.org/77/652877/1/check/openstack-tox-validate/dd70db3/ : FAILURE in 4m 58s\n- releases-tox-list-changes http://logs.openstack.org/77/652877/1/check/releases-tox-list-changes/0be7dde/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c93119470acf664eb1fe801d30b935f5e1c18f3e","author":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"date":"2019-04-16 21:08:12.000000000","message":"Patch Set 1:\n\nBased on the explanation in the commit message, I think we can skip this EM release because they all duplicate an existing release.","accounts_in_message":[],"_revision_number":1},{"id":"f67bea28d5c679308efaba33f138834ae6c39cd0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 00:39:18.000000000","message":"Patch Set 1:\n\n\u003e Based on the explanation in the commit message, I think we can skip\n \u003e this EM release because they all duplicate an existing release.\n\nSorry the point is to add a tag that indicates at which point congress transitions into extended maintenance, to any potential consumers can see clearly in the git history if a patch was gated and released by the congress team.\n\nIt\u0027s a procedural change","accounts_in_message":[],"_revision_number":1},{"id":"33df45d0999cc5da8dfd04a79fc7c14599c78aad","author":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"date":"2019-04-17 18:01:57.000000000","message":"Patch Set 1: Code-Review+1\n\nGot it. LGTM. Assuming the validate will pass once this is in: https://review.openstack.org/#/c/652876/","accounts_in_message":[],"_revision_number":1},{"id":"477a2ae9a2bb249dbbe6779fb70b35fec7270a9b","author":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"date":"2019-04-19 00:36:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9424f09d417c14351bf169003cef156829f792c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 00:53:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/77/652877/2/check/openstack-tox-docs/2879c70/html/ : SUCCESS in 6m 49s\n- openstack-tox-validate http://logs.openstack.org/77/652877/2/check/openstack-tox-validate/f1fcdd2/ : SUCCESS in 4m 14s\n- releases-tox-list-changes http://logs.openstack.org/77/652877/2/check/releases-tox-list-changes/1e6e9c2/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bfbc1e9dc2e0bd8cc7ae00e3308e5d5e17c83f36","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 18:52:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nOne down! Thanks Eric.","accounts_in_message":[],"_revision_number":2},{"id":"43448aebda13c78d1bba51e2ccb54bb9bc8a22e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:52:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a7fbb5850259ac928a358abe370eb839a4e1aa8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:05:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/77/652877/2/gate/openstack-tox-docs/808bcea/html/ : SUCCESS in 7m 01s\n- openstack-tox-validate http://logs.openstack.org/77/652877/2/gate/openstack-tox-validate/a712961/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"140909d42bb7b69a4c354aaaa621080bc8ec6d47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:05:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9d4df9fab1780314995feffe777a2f7eb7343fa8","revisions":{"93e5388fc729cbd01b44036691cd3473cb06b04e":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:57:56.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/77/652877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/652877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/652877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/652877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/652877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/652877/1"}}},"commit":{"parents":[{"commit":"4c5b3ac22aaa3ed8d706a76c8703fef10d166fea","subject":"[congress] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c5b3ac22aaa3ed8d706a76c8703fef10d166fea"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"subject":"[congress] create pike-em tag against final releases","message":"[congress] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Iea97c3534df7d73d2666c4765dfc50eb4c99f306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93e5388fc729cbd01b44036691cd3473cb06b04e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93e5388fc729cbd01b44036691cd3473cb06b04e"}]},"branch":"refs/heads/master"},"9d4df9fab1780314995feffe777a2f7eb7343fa8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-19 00:36:14.000000000","uploader":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"ref":"refs/changes/77/652877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/652877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/652877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/652877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/652877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/652877/2"}}},"commit":{"parents":[{"commit":"821a7313a094d99dd9e5975922678eb41639bb58","subject":"Merge \"Before we checkout a ref try resetting the workdir to HEAD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/821a7313a094d99dd9e5975922678eb41639bb58"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Eric Kao","email":"ekcs.openstack@gmail.com","date":"2019-04-19 00:36:14.000000000","tz":0},"subject":"[congress] create pike-em tag against final releases","message":"[congress] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Iea97c3534df7d73d2666c4765dfc50eb4c99f306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d4df9fab1780314995feffe777a2f7eb7343fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d4df9fab1780314995feffe777a2f7eb7343fa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
