)]}'
{"id":"openstack%2Freleases~652883","triplet_id":"openstack%2Freleases~master~I8b58a6feac1a2d17a3ab6ac1d3daf7e6f25c1c3c","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I8b58a6feac1a2d17a3ab6ac1d3daf7e6f25c1c3c","subject":"[Telemetry] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:58:32.000000000","updated":"2019-05-10 21:31:01.000000000","submitted":"2019-05-10 21:31:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652883-1557523861238-72f24b91","meta_rev_id":"c980cfe3a271c8aae9807218c7df3e6b43ddf520","_number":652883,"virtual_id_number":652883,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-05-10 21:31:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-10 21:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 21:17:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 05:58:32.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-05-10 21:17:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 21:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0df3b1f1fcbc15d32075b19f47318c690a16feb2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:58:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb0a8b99b431bcd639b6428c64dc2465f4b585f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:11:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/652883/1/check/openstack-tox-docs/420f387/html/ : SUCCESS in 7m 42s\n- openstack-tox-validate http://logs.openstack.org/83/652883/1/check/openstack-tox-validate/a6fbb5a/ : FAILURE in 6m 16s\n- releases-tox-list-changes http://logs.openstack.org/83/652883/1/check/releases-tox-list-changes/ac471e0/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"681d6a4a8fa20670ccd552c77c75c437c1a1092b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 06:19:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ab1f0c5af19182e9927e1f915b8f8f5bbe670c88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 06:56:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/652883/2/check/openstack-tox-docs/cb6533d/html/ : SUCCESS in 8m 12s\n- openstack-tox-validate http://logs.openstack.org/83/652883/2/check/openstack-tox-validate/ae40106/ : FAILURE in 5m 50s\n- releases-tox-list-changes http://logs.openstack.org/83/652883/2/check/releases-tox-list-changes/40f1980/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"8d4b962b48e02ca5ed9b9fcd9857d02498bd47f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 21:18:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7ad734c7c59c6062382bd6d9f03339a913b0f41f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 21:30:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/652883/3/check/openstack-tox-docs/8c14928/html/ : SUCCESS in 6m 46s\n- openstack-tox-validate http://logs.openstack.org/83/652883/3/check/openstack-tox-validate/5b50809/ : FAILURE in 4m 01s\n- releases-tox-list-changes http://logs.openstack.org/83/652883/3/check/releases-tox-list-changes/4a3a671/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"99253ee915be6bc12b160a0d37cd663a3835b5d5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 21:09:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"56fee863166951efe9739186500c04a9bcc30971","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:16:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/83/652883/4/check/openstack-tox-docs/4c99293/html/ : SUCCESS in 6m 41s\n- openstack-tox-validate http://logs.openstack.org/83/652883/4/check/openstack-tox-validate/037f308/ : SUCCESS in 4m 08s\n- releases-tox-list-changes http://logs.openstack.org/83/652883/4/check/releases-tox-list-changes/4fd829f/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":4},{"id":"11393c7f37a89ea0821f75caa63a1d795bf29114","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 21:17:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"59ca9dcd43ef4a894577ff5fbbe35bb45925ae94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:17:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4bfff8a06a568263172a81afd876a34e1bbac80f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:31:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cca354f21c43889d1f79918ce4d36a34d94255a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:31:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/83/652883/4/gate/openstack-tox-docs/d9cf155/html/ : SUCCESS in 6m 53s\n- openstack-tox-validate http://logs.openstack.org/83/652883/4/gate/openstack-tox-validate/dc7f5d6/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"07c983cfa483369f11dc282ceebca4ca2fa73f7b","revisions":{"6e8ce02c9b6c70bead89a9110a3691bcc60307c4":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:58:32.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/83/652883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/652883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/652883/1"}}},"commit":{"parents":[{"commit":"b5c21d174a1a696039afac405283c0690cf0dccf","subject":"[Telemetry] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5c21d174a1a696039afac405283c0690cf0dccf"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"subject":"[Telemetry] create pike-em tag against final releases","message":"[Telemetry] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I8b58a6feac1a2d17a3ab6ac1d3daf7e6f25c1c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e8ce02c9b6c70bead89a9110a3691bcc60307c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e8ce02c9b6c70bead89a9110a3691bcc60307c4"}]},"branch":"refs/heads/master"},"c2a846998277fe9f3d7814e663a430f3dbc219bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-17 06:19:40.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/83/652883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/652883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/652883/2"}}},"commit":{"parents":[{"commit":"7f8f60b23e655285d7bc413f67020e212db3b097","subject":"[Telemetry] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7f8f60b23e655285d7bc413f67020e212db3b097"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 06:12:28.000000000","tz":600},"subject":"[Telemetry] create pike-em tag against final releases","message":"[Telemetry] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I8b58a6feac1a2d17a3ab6ac1d3daf7e6f25c1c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c2a846998277fe9f3d7814e663a430f3dbc219bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c2a846998277fe9f3d7814e663a430f3dbc219bc"}]},"branch":"refs/heads/master"},"2d697617846497093c36554c1f1001b4d8f519f0":{"kind":"REWORK","_number":3,"created":"2019-05-09 21:18:04.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/83/652883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/652883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/652883/3"}}},"commit":{"parents":[{"commit":"3716325d254526966c74f3495021a5be86870854","subject":"Merge \"Update cycle process based on PTG review\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3716325d254526966c74f3495021a5be86870854"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-09 21:17:06.000000000","tz":-300},"subject":"[Telemetry] create pike-em tag against final releases","message":"[Telemetry] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended\nmaintenance by creating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there\nare no code changes sine the previous release, sadly that\u0027s\nnon-trivial to detect en masse.)\n\nIf you are the PTL or release liaison for this team please +1 this\nreview if it is ready or -1 and take over the review to correct\nmistakes.\n\nChange-Id: I8b58a6feac1a2d17a3ab6ac1d3daf7e6f25c1c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d697617846497093c36554c1f1001b4d8f519f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d697617846497093c36554c1f1001b4d8f519f0"}]},"branch":"refs/heads/master"},"07c983cfa483369f11dc282ceebca4ca2fa73f7b":{"kind":"REWORK","_number":4,"created":"2019-05-10 21:09:17.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/83/652883/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/652883/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/652883/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/652883/4"}}},"commit":{"parents":[{"commit":"3716325d254526966c74f3495021a5be86870854","subject":"Merge \"Update cycle process based on PTG review\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3716325d254526966c74f3495021a5be86870854"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 21:09:02.000000000","tz":-300},"subject":"[Telemetry] create pike-em tag against final releases","message":"[Telemetry] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended\nmaintenance by creating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there\nare no code changes sine the previous release, sadly that\u0027s\nnon-trivial to detect en masse).\n\nIf you are the PTL or release liaison for this team please +1 this\nreview if it is ready or -1 and take over the review to correct\nmistakes.\n\nChange-Id: I8b58a6feac1a2d17a3ab6ac1d3daf7e6f25c1c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07c983cfa483369f11dc282ceebca4ca2fa73f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07c983cfa483369f11dc282ceebca4ca2fa73f7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
