)]}'
{"id":"openstack%2Freleases~652887","triplet_id":"openstack%2Freleases~master~I5a63924005952f62489b97beb8b89729152ac9bc","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I5a63924005952f62489b97beb8b89729152ac9bc","subject":"[OpenStackSDK] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:58:55.000000000","updated":"2019-04-22 22:05:24.000000000","submitted":"2019-04-22 22:05:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652887-1555970724149-81340386","meta_rev_id":"0c280ac06851f24bf7261a62f083a1542bf0e8d9","_number":652887,"virtual_id_number":652887,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-22 22:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-22 21:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-04-22 21:54:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 05:58:55.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-22 21:54:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-22 22:05:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58dd376f51deaa2f167f380df285bfbea064ec19","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:58:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a14abf89e79437d8f6e885fa519ece9290859e66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:12:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/87/652887/1/check/openstack-tox-docs/1c866db/html/ : SUCCESS in 7m 56s\n- openstack-tox-validate http://logs.openstack.org/87/652887/1/check/openstack-tox-validate/6174f66/ : FAILURE in 5m 12s\n- releases-tox-list-changes http://logs.openstack.org/87/652887/1/check/releases-tox-list-changes/c92a49d/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5bcba862d4f1b0f8438866c01d100f2dd801bbd3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 06:07:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aea0aab6c33854e47e8c9fae7c36a5bbf93108fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 06:36:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/87/652887/2/check/openstack-tox-docs/599ed59/html/ : SUCCESS in 7m 05s\n- openstack-tox-validate http://logs.openstack.org/87/652887/2/check/openstack-tox-validate/2beee39/ : FAILURE in 5m 00s\n- releases-tox-list-changes http://logs.openstack.org/87/652887/2/check/releases-tox-list-changes/db53957/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"200dc0a5214b4b5946c7aba8dc90ebc4140a7410","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:38:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"62aed294ceb5cf250ea805cbe93b7e3eb18712fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:51:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/652887/2/check/openstack-tox-docs/5f322c3/html/ : SUCCESS in 7m 17s\n- openstack-tox-validate http://logs.openstack.org/87/652887/2/check/openstack-tox-validate/27a1937/ : SUCCESS in 4m 10s\n- releases-tox-list-changes http://logs.openstack.org/87/652887/2/check/releases-tox-list-changes/137c443/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"703ea5c01c893a64045afbce8d01341b460d8088","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:54:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b2b08d097451c1103ac0bcb49893792806d029e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:54:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be09cefd2c610fbc3ddf8a7793574b7090f5b2bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:55:08.000000000","message":"Patch Set 2:\n\nShoot, might need to update now with the namespace migrations.","accounts_in_message":[],"_revision_number":2},{"id":"f1e5f5accfaf450b80703a7c6cc84f83347e33d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:05:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/652887/2/gate/openstack-tox-docs/563f701/html/ : SUCCESS in 7m 16s\n- openstack-tox-validate http://logs.openstack.org/87/652887/2/gate/openstack-tox-validate/80a278e/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"789490b010d5e4ed1f06595a1b37fb83ffbf4988","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:05:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3420aee3bd8a520bf6b0dbfc7ada1bb94bf6d476","revisions":{"8ae9ab4ab0b1eab63ec42c98451c5cc6db123e53":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:58:55.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/652887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/652887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/652887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/652887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/652887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/652887/1"}}},"commit":{"parents":[{"commit":"982058821a555905a9cb5c26142507039ae1a3db","subject":"[OpenStackSDK] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/982058821a555905a9cb5c26142507039ae1a3db"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"subject":"[OpenStackSDK] create pike-em tag against final releases","message":"[OpenStackSDK] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I5a63924005952f62489b97beb8b89729152ac9bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ae9ab4ab0b1eab63ec42c98451c5cc6db123e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ae9ab4ab0b1eab63ec42c98451c5cc6db123e53"}]},"branch":"refs/heads/master"},"3420aee3bd8a520bf6b0dbfc7ada1bb94bf6d476":{"kind":"REWORK","_number":2,"created":"2019-04-17 06:07:58.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/652887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/652887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/652887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/652887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/652887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/652887/2"}}},"commit":{"parents":[{"commit":"1262e7e31037a27c8a0bbc37b017488ae9dba5ae","subject":"[OpenStackSDK] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1262e7e31037a27c8a0bbc37b017488ae9dba5ae"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 06:07:48.000000000","tz":600},"subject":"[OpenStackSDK] create pike-em tag against final releases","message":"[OpenStackSDK] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I5a63924005952f62489b97beb8b89729152ac9bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3420aee3bd8a520bf6b0dbfc7ada1bb94bf6d476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3420aee3bd8a520bf6b0dbfc7ada1bb94bf6d476"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
