)]}'
{"id":"openstack%2Freleases~652889","triplet_id":"openstack%2Freleases~master~I1256efb1059029ae2e901a1d18ea445a91bd1a3a","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I1256efb1059029ae2e901a1d18ea445a91bd1a3a","subject":"[oslo] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:59:07.000000000","updated":"2019-04-22 22:05:21.000000000","submitted":"2019-04-22 22:05:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652889-1555970721468-81340386","meta_rev_id":"325e7b2c19f96b0dbf03e6c95902c8c1ae1bcef5","_number":652889,"virtual_id_number":652889,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-04-16 22:58:19.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-04-22 22:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-04-17 15:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-04-22 21:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-04-18 09:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-04-22 21:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 15:36:32.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-04-18 09:17:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-04-22 21:52:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-22 22:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c78ff42168f0232cd31a118ab1cc328289fef275","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:59:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ca33a20776db93d3259723a7293631682847a71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 07:54:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/652889/1/check/openstack-tox-docs/d9dd8a3/html/ : SUCCESS in 7m 04s\n- openstack-tox-validate http://logs.openstack.org/89/652889/1/check/openstack-tox-validate/a09df42/ : FAILURE in 8m 31s\n- releases-tox-list-changes http://logs.openstack.org/89/652889/1/check/releases-tox-list-changes/0e6a432/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1d471155bd2a605db0d0614f2d6ac2861492d59d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-16 15:45:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77691b837a6390ba232b0a1774e2b836269fa9f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 16:05:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/652889/2/check/openstack-tox-docs/82cac45/html/ : SUCCESS in 7m 32s\n- openstack-tox-validate http://logs.openstack.org/89/652889/2/check/openstack-tox-validate/5cd3659/ : FAILURE in 8m 35s\n- releases-tox-list-changes http://logs.openstack.org/89/652889/2/check/releases-tox-list-changes/e2a4885/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2fb6808c8d69c2e1bc726a4e2b85798611023046","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-16 16:25:22.000000000","message":"Patch Set 2: Code-Review-1\n\nThis appears to be very broken. Is it a known problem? I\u0027d rather not try to fix all of these manually.","accounts_in_message":[],"_revision_number":2},{"id":"d3c2336778a76aae11a416fc8c5fd91c8ea35541","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 22:57:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a8463a8f5a9eb94a06688d34644f89ad21d1fb5c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 22:58:19.000000000","message":"Patch Set 2:\n\n\u003e This appears to be very broken. Is it a known problem? I\u0027d rather\n \u003e not try to fix all of these manually.\n\nI think it just *looks* broken because we can\u0027t do stacked changes for releases and the \u0027create final releases\u0027 change hadn\u0027t merged.\n\nI\u0027m pretty sure this will come back +1 from Zuul now that it has.","accounts_in_message":[],"_revision_number":2},{"id":"2a0091310f1365d4e49d284c04ed75d8906413a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 23:07:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/652889/2/check/openstack-tox-docs/e24c900/html/ : SUCCESS in 8m 07s\n- openstack-tox-validate http://logs.openstack.org/89/652889/2/check/openstack-tox-validate/7c00a9b/ : SUCCESS in 5m 03s\n- releases-tox-list-changes http://logs.openstack.org/89/652889/2/check/releases-tox-list-changes/17a912b/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"50d22d125e7a7c2fa783f0688b91e617e5b3530a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-17 15:36:32.000000000","message":"Patch Set 2: Code-Review+1\n\nOkay, cool. lgtm.","accounts_in_message":[],"_revision_number":2},{"id":"19f7ecf172f07c96aa472e214c5d00736551a89a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-18 09:17:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"253f34a424c7cc7df158620bc4bd31f688b1a641","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:52:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba1ccd91fe932674cad34f9138ce57aec7aed5f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:53:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b08ee459501d442f94de9dd1660a7aa3f159860a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:05:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/652889/2/gate/openstack-tox-docs/2dd70e9/html/ : SUCCESS in 6m 46s\n- openstack-tox-validate http://logs.openstack.org/89/652889/2/gate/openstack-tox-validate/7678ec9/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"08128587b08520c5f5dc4111b85aed24f4b43476","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:05:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fcc47dc73b47d7b0dde59135282aafa8dd1b3fd7","revisions":{"6136d75d7c7071eedd5d984c35912c92715430f0":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:59:07.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/89/652889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/652889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/652889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/652889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/652889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/652889/1"}}},"commit":{"parents":[{"commit":"bdb86c734e7b81a4ff490e200ee7e20e4ee2a41b","subject":"[oslo] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bdb86c734e7b81a4ff490e200ee7e20e4ee2a41b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"subject":"[oslo] create pike-em tag against final releases","message":"[oslo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I1256efb1059029ae2e901a1d18ea445a91bd1a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6136d75d7c7071eedd5d984c35912c92715430f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6136d75d7c7071eedd5d984c35912c92715430f0"}]},"branch":"refs/heads/master"},"fcc47dc73b47d7b0dde59135282aafa8dd1b3fd7":{"kind":"REWORK","_number":2,"created":"2019-04-16 15:45:00.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/89/652889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/652889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/652889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/652889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/652889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/652889/2"}}},"commit":{"parents":[{"commit":"f713cf7852acb1e456f3f2a081750b41feb8cfdb","subject":"[oslo] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f713cf7852acb1e456f3f2a081750b41feb8cfdb"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-16 15:44:44.000000000","tz":0},"subject":"[oslo] create pike-em tag against final releases","message":"[oslo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I1256efb1059029ae2e901a1d18ea445a91bd1a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fcc47dc73b47d7b0dde59135282aafa8dd1b3fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fcc47dc73b47d7b0dde59135282aafa8dd1b3fd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
