)]}'
{"id":"openstack%2Freleases~652891","triplet_id":"openstack%2Freleases~master~Ibe39ae9da829290cffb6ab743144f58c7193d7c7","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"Ibe39ae9da829290cffb6ab743144f58c7193d7c7","subject":"[sahara] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:59:19.000000000","updated":"2019-05-10 21:42:42.000000000","submitted":"2019-05-10 21:42:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652891-1557524562228-f41f2753","meta_rev_id":"2be94b706371a3dd53f4505642113622172f00e7","_number":652891,"virtual_id_number":652891,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2019-05-10 21:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-10 21:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 21:30:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 05:59:19.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-05-10 21:30:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 21:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41b6eb2535f4dc15dbd0dadab883a6782d4e6f27","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f7c1c6f4a533bdc507caa6c1238ffba36efaade","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 07:55:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/652891/1/check/openstack-tox-docs/bce34e1/html/ : SUCCESS in 7m 06s\n- openstack-tox-validate http://logs.openstack.org/91/652891/1/check/openstack-tox-validate/2d70397/ : FAILURE in 5m 21s\n- releases-tox-list-changes http://logs.openstack.org/91/652891/1/check/releases-tox-list-changes/d9b28f8/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5156cb3fc48a512dd87cf224061c3441a203e98b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-18 05:46:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d95d53716ce44671d26ef66c8dd17b12e8dc234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 06:00:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/652891/1/check/openstack-tox-docs/39bbe1a/html/ : SUCCESS in 7m 18s\n- openstack-tox-validate http://logs.openstack.org/91/652891/1/check/openstack-tox-validate/76db873/ : FAILURE in 4m 20s\n- releases-tox-list-changes http://logs.openstack.org/91/652891/1/check/releases-tox-list-changes/f8555d6/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"773cf1b5c671142385db2029a0b5512d63260693","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:49:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"017ecb624333e8e4f36702df1d55d0f8d3e4fa27","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:51:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df6fe45214103ffa4affa6576466f8610b31d7a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:05:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/652891/3/check/openstack-tox-docs/1f0002c/html/ : SUCCESS in 7m 30s\n- openstack-tox-validate http://logs.openstack.org/91/652891/3/check/openstack-tox-validate/1dc495c/ : FAILURE in 4m 22s\n- releases-tox-list-changes http://logs.openstack.org/91/652891/3/check/releases-tox-list-changes/fb789fc/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ea3bea4bd95981d5e24c70a1d1cad083faa9ad66","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-26 16:51:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b65138dcefbac765b9206174c022e2c3b3e1c39d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:22:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/652891/3/check/openstack-tox-docs/0880785/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/91/652891/3/check/openstack-tox-validate/3110829/ : FAILURE in 3m 50s\n- releases-tox-list-changes http://logs.openstack.org/91/652891/3/check/releases-tox-list-changes/35bede4/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"344befb5f6f26bb9984c90ca9e4215dd20534855","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 20:40:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9d36f99472561c1b89fbc3ca4c7ce8d359108fba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 20:54:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/652891/4/check/openstack-tox-docs/8dc7067/html/ : SUCCESS in 7m 09s\n- openstack-tox-validate http://logs.openstack.org/91/652891/4/check/openstack-tox-validate/f9f88ad/ : FAILURE in 4m 07s\n- releases-tox-list-changes http://logs.openstack.org/91/652891/4/check/releases-tox-list-changes/a821b95/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"ad5a1bade42890086ce2d242afbc3fc7b453838b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 21:20:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d3671dc23323cc6a6d59edc242b4038a0f73e4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:29:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/652891/5/check/openstack-tox-docs/cd04956/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/91/652891/5/check/openstack-tox-validate/5e8b386/ : SUCCESS in 4m 11s\n- releases-tox-list-changes http://logs.openstack.org/91/652891/5/check/releases-tox-list-changes/9f62d05/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":5},{"id":"348993aa3f5bee21d63b657169c5914c6852a8fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 21:30:20.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5ba6141e8192518b212a5fdd8586f94a75e53e5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:30:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d7d5f1b6787e731f72cea6ca3bf9e5951d069a80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:42:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/652891/5/gate/openstack-tox-docs/68cef67/html/ : SUCCESS in 7m 12s\n- openstack-tox-validate http://logs.openstack.org/91/652891/5/gate/openstack-tox-validate/6a91daa/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"6ecb54c0e8495f2fde268cae851e5d72e0b42942","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:42:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"54386155adf10f9018da6478ba752ce11f6185a7","revisions":{"5211dff4f87bb835e629c015e3a47dfd231ebf10":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:59:19.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/91/652891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/652891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/652891/1"}}},"commit":{"parents":[{"commit":"2ae76a396240a080ad40486696409a3c8db117da","subject":"[sahara] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ae76a396240a080ad40486696409a3c8db117da"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"subject":"[sahara] create pike-em tag against final releases","message":"[sahara] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ibe39ae9da829290cffb6ab743144f58c7193d7c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5211dff4f87bb835e629c015e3a47dfd231ebf10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5211dff4f87bb835e629c015e3a47dfd231ebf10"}]},"branch":"refs/heads/master"},"21b920dd5126ba801286e688bb81cbf07a257688":{"kind":"REWORK","_number":2,"created":"2019-04-22 21:49:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/652891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/652891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/652891/2"}}},"commit":{"parents":[{"commit":"a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d","subject":"Merge \"[ironic] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 21:49:47.000000000","tz":-300},"subject":"[sahara] create pike-em tag against final releases","message":"[sahara] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended\nmaintenance by creating the pike-em tag.\n\nChange-Id: Ibe39ae9da829290cffb6ab743144f58c7193d7c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21b920dd5126ba801286e688bb81cbf07a257688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21b920dd5126ba801286e688bb81cbf07a257688"}]},"branch":"refs/heads/master"},"46742d7095c4019f3f1f74148ce24b76af780240":{"kind":"REWORK","_number":3,"created":"2019-04-22 21:51:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/652891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/652891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/652891/3"}}},"commit":{"parents":[{"commit":"a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d","subject":"Merge \"[ironic] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 21:51:50.000000000","tz":-300},"subject":"[sahara] create pike-em tag against final releases","message":"[sahara] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended\nmaintenance by creating the pike-em tag.\n\nChange-Id: Ibe39ae9da829290cffb6ab743144f58c7193d7c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46742d7095c4019f3f1f74148ce24b76af780240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46742d7095c4019f3f1f74148ce24b76af780240"}]},"branch":"refs/heads/master"},"c05161e04389aeee5268904c11d6a46d85508f16":{"kind":"REWORK","_number":4,"created":"2019-05-10 20:40:39.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/652891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/652891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/652891/4"}}},"commit":{"parents":[{"commit":"a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d","subject":"Merge \"[ironic] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 20:39:46.000000000","tz":-300},"subject":"[sahara] create pike-em tag against final releases","message":"[sahara] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended\nmaintenance by creating the pike-em tag.\n\nChange-Id: Ibe39ae9da829290cffb6ab743144f58c7193d7c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c05161e04389aeee5268904c11d6a46d85508f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c05161e04389aeee5268904c11d6a46d85508f16"}]},"branch":"refs/heads/master"},"54386155adf10f9018da6478ba752ce11f6185a7":{"kind":"REWORK","_number":5,"created":"2019-05-10 21:20:55.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/652891/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/652891/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/652891/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/652891/5"}}},"commit":{"parents":[{"commit":"a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d","subject":"Merge \"[ironic] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a539f1df4e8eccb4e2061e3dc8c1d09c2e14775d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 21:20:49.000000000","tz":-300},"subject":"[sahara] create pike-em tag against final releases","message":"[sahara] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended\nmaintenance by creating the pike-em tag.\n\nChange-Id: Ibe39ae9da829290cffb6ab743144f58c7193d7c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54386155adf10f9018da6478ba752ce11f6185a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54386155adf10f9018da6478ba752ce11f6185a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
