)]}'
{"id":"openstack%2Freleases~652895","triplet_id":"openstack%2Freleases~master~Id204b5d1aa085bbde5aa7c94c85d221eb1ee0329","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"Id204b5d1aa085bbde5aa7c94c85d221eb1ee0329","subject":"[blazar] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:59:43.000000000","updated":"2019-05-10 16:04:07.000000000","submitted":"2019-05-10 16:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652895-1557504247536-71816a1a","meta_rev_id":"280091274741abdcd77a08b67602a06f79bd3798","_number":652895,"virtual_id_number":652895,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-10 16:04:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-09 21:02:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-10 15:17:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 15:23:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 15:17:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-05-10 15:23:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 16:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23a4ee9b5226446f912fb48b4ada7e916a0bc1fb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:59:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"062949f912e6bbb288c5de6ed43eee414b8df72a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 08:07:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/652895/1/check/openstack-tox-docs/4ffa137/html/ : SUCCESS in 6m 56s\n- openstack-tox-validate http://logs.openstack.org/95/652895/1/check/openstack-tox-validate/f92c9ba/ : FAILURE in 4m 01s\n- releases-tox-list-changes http://logs.openstack.org/95/652895/1/check/releases-tox-list-changes/8c10747/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4ab42d8f1f19b4d0aae3ca3ff78c19cab0cabf50","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 20:28:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"655935fa153e86304b61c2d500019577057c5d90","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 20:31:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"588a3df736545b40270763702721edb105cb0702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:46:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/652895/3/check/openstack-tox-docs/98b4eb2/html/ : SUCCESS in 7m 18s\n- openstack-tox-validate http://logs.openstack.org/95/652895/3/check/openstack-tox-validate/42c1565/ : SUCCESS in 4m 05s\n- releases-tox-list-changes http://logs.openstack.org/95/652895/3/check/releases-tox-list-changes/5e5fa28/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b587f272cc1f3f0304522c296f5c0cffb5bce5f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 21:02:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9ae8d8a1d8ce77957318054247268aa0fb5c18bb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-10 15:17:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"629f2d20d5950f61ad6c9ed43663c13f3f0e2137","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 15:23:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9b69c9c62f0956e79ffd7bed4b3d8dc5207f64bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:23:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"887fb61be3d076197a2791671c7771eab4ecefb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:32:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/652895/3/gate/openstack-tox-docs/c73f4e0/ : POST_FAILURE in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/95/652895/3/gate/openstack-tox-validate/d2c4d62/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"c3a794ce28a4241fc3636db2948a7f7a7ed4628b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 15:46:09.000000000","message":"Patch Set 3:\n\nrecheck\n\nUnable to reproduce failure locally.","accounts_in_message":[],"_revision_number":3},{"id":"1a3a223a61995348299f8f6885feff4e4a32ab7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:55:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/652895/3/check/openstack-tox-docs/37d5073/html/ : SUCCESS in 6m 13s\n- openstack-tox-validate http://logs.openstack.org/95/652895/3/check/openstack-tox-validate/bcca06b/ : SUCCESS in 3m 34s\n- releases-tox-list-changes http://logs.openstack.org/95/652895/3/check/releases-tox-list-changes/ef3f43b/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"5b80ba196aa6dfc815b1d62165fe07650f20abd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:56:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"18d38e102dbe80e90ff604c2f06a0068fffa6278","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:04:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/652895/3/gate/openstack-tox-docs/7e137b5/html/ : SUCCESS in 6m 06s\n- openstack-tox-validate http://logs.openstack.org/95/652895/3/gate/openstack-tox-validate/2d51263/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"f698b7b6a05da248e1a6bcbcc7d4614dd4abde78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:04:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"71d80836276befdc6e3631f3d2795013fb5324d5","revisions":{"b5e5504e5238a5663c3323f1235132c14b60084d":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:59:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/95/652895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/652895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/652895/1"}}},"commit":{"parents":[{"commit":"76a67b0907e59880834cc700497c98bcd215a181","subject":"[blazar] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76a67b0907e59880834cc700497c98bcd215a181"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"subject":"[blazar] create pike-em tag against final releases","message":"[blazar] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Id204b5d1aa085bbde5aa7c94c85d221eb1ee0329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5e5504e5238a5663c3323f1235132c14b60084d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5e5504e5238a5663c3323f1235132c14b60084d"}]},"branch":"refs/heads/master"},"8009a3811c5643d8a0cfb1724e6df493d220a823":{"kind":"REWORK","_number":2,"created":"2019-05-09 20:28:03.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/95/652895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/652895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/652895/2"}}},"commit":{"parents":[{"commit":"76a67b0907e59880834cc700497c98bcd215a181","subject":"[blazar] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76a67b0907e59880834cc700497c98bcd215a181"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 20:28:01.000000000","tz":0},"subject":"[blazar] create pike-em tag against final releases","message":"[blazar] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Id204b5d1aa085bbde5aa7c94c85d221eb1ee0329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8009a3811c5643d8a0cfb1724e6df493d220a823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8009a3811c5643d8a0cfb1724e6df493d220a823"}]},"branch":"refs/heads/master"},"71d80836276befdc6e3631f3d2795013fb5324d5":{"kind":"REWORK","_number":3,"created":"2019-05-09 20:31:38.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/95/652895/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/652895/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/652895/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/652895/3"}}},"commit":{"parents":[{"commit":"03ccc759a8a99874e983e4b8b1ef5161d22ebd3c","subject":"Merge \"oslo.db 4.40.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/03ccc759a8a99874e983e4b8b1ef5161d22ebd3c"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 20:31:21.000000000","tz":-240},"subject":"[blazar] create pike-em tag against final releases","message":"[blazar] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Id204b5d1aa085bbde5aa7c94c85d221eb1ee0329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71d80836276befdc6e3631f3d2795013fb5324d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71d80836276befdc6e3631f3d2795013fb5324d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
