)]}'
{"id":"openstack%2Freleases~652900","triplet_id":"openstack%2Freleases~master~I4a12cf4068b2c13a1ef448dfc1083f52f38baaf3","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I4a12cf4068b2c13a1ef448dfc1083f52f38baaf3","subject":"[neutron] final releases for pike","status":"MERGED","created":"2019-04-16 06:00:18.000000000","updated":"2019-05-10 19:44:13.000000000","submitted":"2019-05-10 19:44:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652900-1557517453450-72f24b91","meta_rev_id":"949984ba14f9c1d467d67abaeaf62ad8603ac9a6","_number":652900,"virtual_id_number":652900,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-05-10 19:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-10 19:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 19:28:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 06:00:18.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-04-27 14:20:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-04-27 14:21:08.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-05-10 19:28:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 19:44:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51908d01dbd7ad0a6323c6bda25d1087e9b8bb14","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 06:00:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9427e8ae1f7f41c6b104c38cb766f39481069732","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 08:29:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/652900/1/check/openstack-tox-docs/0c1c7a8/html/ : SUCCESS in 7m 41s\n- openstack-tox-validate http://logs.openstack.org/00/652900/1/check/openstack-tox-validate/816d435/ : FAILURE in 10m 33s\n- releases-tox-list-changes http://logs.openstack.org/00/652900/1/check/releases-tox-list-changes/47b2709/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"00c771d81eb93ae518ec44907670bef52f5cb39b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 06:43:50.000000000","message":"Patch Set 1:\n\nNeeds to be updated once https://review.openstack.org/#/c/653304/ merges","accounts_in_message":[],"_revision_number":1},{"id":"06f244ecec79acfc57bf61998da4e05955c6462b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:14:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4fd8a2b5999e1372e5d2138836aa8b94bcdbcd3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:32:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/652900/2/check/openstack-tox-docs/bf4e1c9/html/ : SUCCESS in 7m 03s\n- openstack-tox-validate http://logs.openstack.org/00/652900/2/check/openstack-tox-validate/f9db58b/ : FAILURE in 11m 13s\n- releases-tox-list-changes http://logs.openstack.org/00/652900/2/check/releases-tox-list-changes/1ae898a/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"28f80e597d599ee1d492c371a95e893af28b5b95","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-27 13:28:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fd1f45325ca2e63db2b068b59f52195d4ef019ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 13:39:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/652900/3/check/openstack-tox-docs/1fc050b/html/ : SUCCESS in 7m 19s\n- openstack-tox-validate http://logs.openstack.org/00/652900/3/check/openstack-tox-validate/c52c085/ : FAILURE in 9m 16s\n- releases-tox-list-changes http://logs.openstack.org/00/652900/3/check/releases-tox-list-changes/c1e701e/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"f8abbb9739bfa016c7cd95fa127f84335941c29d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-27 14:20:59.000000000","message":"Patch Set 3:\n\n\u003e Needs to be updated once https://review.openstack.org/#/c/653304/\n \u003e merges\n\nI suppose we need a networking-bagpipe pike release that includes this fix?","accounts_in_message":[],"_revision_number":3},{"id":"2fe8111b96aa6c05a8cf9b5973023213b18eff9e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 19:03:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"12b1fd066a1f79d9bf2f7cd58abc5969ae3a1cd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:19:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/652900/4/check/openstack-tox-docs/9b44b75/html/ : SUCCESS in 7m 26s\n- openstack-tox-validate http://logs.openstack.org/00/652900/4/check/openstack-tox-validate/a683096/ : SUCCESS in 10m 29s\n- releases-tox-list-changes http://logs.openstack.org/00/652900/4/check/releases-tox-list-changes/cc32cad/ : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":4},{"id":"6b8bbe134d1bcb74c84e8abd9b6722740726aa23","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 19:28:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aa8b75fb4e8e14230d8eb35361af009140d8621d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:28:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1abcf9431fbb370e83d52299e3f97a2882dc3714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:44:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"83a216cbbb3608161bf7465786a9e5d09eb0c3d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:44:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/652900/4/gate/openstack-tox-docs/1a133f4/html/ : SUCCESS in 6m 55s\n- openstack-tox-validate http://logs.openstack.org/00/652900/4/gate/openstack-tox-validate/e10f8c0/ : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e350c8b97157fbcea30ea36a273afc07d9a4da0b","revisions":{"7b0754a68daeafb33a8a888fee51b1dd753beac6":{"kind":"REWORK","_number":1,"created":"2019-04-16 06:00:18.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/00/652900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/652900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/652900/1"}}},"commit":{"parents":[{"commit":"a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df","subject":"Merge \"Branch roles that were added during stein cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:14.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:14.000000000","tz":600},"subject":"[neutron] final releases for pike","message":"[neutron] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I4a12cf4068b2c13a1ef448dfc1083f52f38baaf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b0754a68daeafb33a8a888fee51b1dd753beac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b0754a68daeafb33a8a888fee51b1dd753beac6"}]},"branch":"refs/heads/master"},"14d505a42f0f9ef047cfb217b5ee8d97d63af972":{"kind":"NO_CHANGE","_number":2,"created":"2019-04-22 21:14:58.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/00/652900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/652900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/652900/2"}}},"commit":{"parents":[{"commit":"a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df","subject":"Merge \"Branch roles that were added during stein cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:14.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 21:14:50.000000000","tz":-300},"subject":"[neutron] final releases for pike","message":"[neutron] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I4a12cf4068b2c13a1ef448dfc1083f52f38baaf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d505a42f0f9ef047cfb217b5ee8d97d63af972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d505a42f0f9ef047cfb217b5ee8d97d63af972"}]},"branch":"refs/heads/master"},"ae5e1aba2cc109d08f9ad88b130dcc1068e57cfe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-27 13:28:15.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/00/652900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/652900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/652900/3"}}},"commit":{"parents":[{"commit":"4936cae262e348c33f8844319421233acc2158a0","subject":"Merge \"[ironic] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4936cae262e348c33f8844319421233acc2158a0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:14.000000000","tz":600},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-27 13:28:15.000000000","tz":0},"subject":"[neutron] final releases for pike","message":"[neutron] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I4a12cf4068b2c13a1ef448dfc1083f52f38baaf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae5e1aba2cc109d08f9ad88b130dcc1068e57cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae5e1aba2cc109d08f9ad88b130dcc1068e57cfe"}]},"branch":"refs/heads/master"},"e350c8b97157fbcea30ea36a273afc07d9a4da0b":{"kind":"REWORK","_number":4,"created":"2019-05-10 19:03:43.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/00/652900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/652900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/652900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/652900/4"}}},"commit":{"parents":[{"commit":"4936cae262e348c33f8844319421233acc2158a0","subject":"Merge \"[ironic] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4936cae262e348c33f8844319421233acc2158a0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:14.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 19:01:43.000000000","tz":-300},"subject":"[neutron] final releases for pike","message":"[neutron] final releases for pike\n\nThis is a procedural review to tag a final release for each\ndeliverable.\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are\nno code changes sine the previous release, sadly that\u0027s non-trivial to\ndetect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this\nreview if it is ready or -1 and take over the review to correct\nmistakes.\n\nChange-Id: I4a12cf4068b2c13a1ef448dfc1083f52f38baaf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e350c8b97157fbcea30ea36a273afc07d9a4da0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e350c8b97157fbcea30ea36a273afc07d9a4da0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
