)]}'
{"id":"openstack%2Freleases~652914","triplet_id":"openstack%2Freleases~master~I7efab90fc9ebb7e4bc3f0e59f07c51a972414062","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I7efab90fc9ebb7e4bc3f0e59f07c51a972414062","subject":"[Puppet OpenStack] final releases for pike","status":"MERGED","created":"2019-04-16 06:01:53.000000000","updated":"2019-05-16 22:59:40.000000000","submitted":"2019-05-16 22:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652914-1558047580951-deb1f29f","meta_rev_id":"eafa3410c91a8d9e05dc0d491d24df836f0c933c","_number":652914,"virtual_id_number":652914,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-05-15 09:38:51.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-16 22:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-16 22:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-05-16 21:50:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-16 22:48:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 03:09:58.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-05-15 09:38:51.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-16 21:50:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-16 22:48:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-16 22:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a925456f147126547d6bf37835c9a94ec78ef3c9","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 06:01:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"893a13eb908aa2dbdbd0ac6aac99309bd1b5fa60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 09:16:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/14/652914/1/check/openstack-tox-docs/430afb6/html/ : SUCCESS in 6m 46s\n- openstack-tox-validate http://logs.openstack.org/14/652914/1/check/openstack-tox-validate/6ef80e6/ : FAILURE in 7m 00s\n- releases-tox-list-changes http://logs.openstack.org/14/652914/1/check/releases-tox-list-changes/e91ef16/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6d0f438acda8e8bf46f5a767fb18542b553965db","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-17 03:09:58.000000000","message":"Patch Set 1:\n\nWaiting for these patches[1] to merge. Those hash values should be changed. All puppet module versions are unified to 11.6.0, except puppet-vswitch（7.5.0）, puppet-vitrage（1.2.0）\n[1]https://review.openstack.org/#/q/topic:pre-release-pike-final+(status:open+OR+status:merged)\n\nAll","accounts_in_message":[],"_revision_number":1},{"id":"1b08a20d83eed08e888dc87f38dbd5f2d1ac8e8c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-19 00:02:59.000000000","message":"Patch Set 1:\n\nLooks like the referenced patches have all merged. Do you have updated hashes to use zhongshenping?","accounts_in_message":[],"_revision_number":1},{"id":"b67cde692b823f90e73fbe75ba7658a6bd727965","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-19 02:00:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5f3ed58571aefd243d13c88791afce513f819bad","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-19 02:36:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4754c42b4725e5d7daf62d719eb7bf064da83d48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 03:30:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/14/652914/3/check/openstack-tox-docs/0f630b5/html/ : SUCCESS in 6m 49s\n- openstack-tox-validate http://logs.openstack.org/14/652914/3/check/openstack-tox-validate/2238cd4/ : FAILURE in 7m 31s\n- releases-tox-list-changes http://logs.openstack.org/14/652914/3/check/releases-tox-list-changes/33e133f/ : FAILURE in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"a27852766f3da819d57066fbd291cb33f3f84bd2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-19 03:43:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"04b0498e54b9e9c39ee3b4e6a0dab0d7821b0634","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 04:01:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/652914/4/check/openstack-tox-docs/e381c35/html/ : SUCCESS in 7m 08s\n- openstack-tox-validate http://logs.openstack.org/14/652914/4/check/openstack-tox-validate/9509868/ : SUCCESS in 7m 31s\n- releases-tox-list-changes http://logs.openstack.org/14/652914/4/check/releases-tox-list-changes/6a15b67/ : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":4},{"id":"a61a46afd962d67948f2e919ddd534709a1dabae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 18:52:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d62b5ee8ec003f220ef1fcc7c1cd627d2fbb015d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 16:15:42.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"439e2df3663e3070da6fbb8cc2c47b33ddf02f2f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-15 09:38:51.000000000","message":"Patch Set 4:\n\nWaiting for Sean\u0027s comment to be addressed.","accounts_in_message":[],"_revision_number":4},{"id":"dcba8ffed77cb8fef04a119adab04faa501a94ff","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-16 21:50:38.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7184014e0744f3fc6ec1c1279b259d61c82f23b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 22:40:30.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks, makes sense.","accounts_in_message":[],"_revision_number":4},{"id":"a165029ae5c5dc1c696c213557f42bdaf521bef0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 22:48:09.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"86af3f40088d7f3c3fa5c637705861abe1dd5085","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:48:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6fec03bd1495f54bc50b3a701503607dddc2ec30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:59:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/652914/4/gate/openstack-tox-docs/c460084/html/ : SUCCESS in 6m 31s\n- openstack-tox-validate http://logs.openstack.org/14/652914/4/gate/openstack-tox-validate/4fe2f08/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":4},{"id":"42a0d8c1181f3d41e2f976220b8d1d6ed9742d61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:59:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2c2badcf773f2e6bfbfa1457bbf2c515949efcbf","revisions":{"9b1f117c5f07d88623cccea8f0c079a0e8fe286a":{"kind":"REWORK","_number":1,"created":"2019-04-16 06:01:53.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/14/652914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/652914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/652914/1"}}},"commit":{"parents":[{"commit":"a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df","subject":"Merge \"Branch roles that were added during stein cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:16.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:16.000000000","tz":600},"subject":"[Puppet OpenStack] final releases for pike","message":"[Puppet OpenStack] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I7efab90fc9ebb7e4bc3f0e59f07c51a972414062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b1f117c5f07d88623cccea8f0c079a0e8fe286a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b1f117c5f07d88623cccea8f0c079a0e8fe286a"}]},"branch":"refs/heads/master"},"10822151ac7d3170896117dd1af1fe2ea6cb47a5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-19 02:00:31.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/14/652914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/652914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/652914/2"}}},"commit":{"parents":[{"commit":"326374236b856511450406508f5276595d024ff7","subject":"Merge \"[OpenStackSDK] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/326374236b856511450406508f5276595d024ff7"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:16.000000000","tz":600},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-19 02:00:21.000000000","tz":480},"subject":"[Puppet OpenStack] final releases for pike","message":"[Puppet OpenStack] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I7efab90fc9ebb7e4bc3f0e59f07c51a972414062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/10822151ac7d3170896117dd1af1fe2ea6cb47a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/10822151ac7d3170896117dd1af1fe2ea6cb47a5"}]},"branch":"refs/heads/master"},"5611cd5fa3f0a9dce2795bb887becae801c662fa":{"kind":"REWORK","_number":3,"created":"2019-04-19 02:36:10.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/14/652914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/652914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/652914/3"}}},"commit":{"parents":[{"commit":"326374236b856511450406508f5276595d024ff7","subject":"Merge \"[OpenStackSDK] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/326374236b856511450406508f5276595d024ff7"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:16.000000000","tz":600},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-19 02:35:05.000000000","tz":480},"subject":"[Puppet OpenStack] final releases for pike","message":"[Puppet OpenStack] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I7efab90fc9ebb7e4bc3f0e59f07c51a972414062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5611cd5fa3f0a9dce2795bb887becae801c662fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5611cd5fa3f0a9dce2795bb887becae801c662fa"}]},"branch":"refs/heads/master"},"2c2badcf773f2e6bfbfa1457bbf2c515949efcbf":{"kind":"REWORK","_number":4,"created":"2019-04-19 03:43:50.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/14/652914/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/652914/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/652914/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/652914/4"}}},"commit":{"parents":[{"commit":"326374236b856511450406508f5276595d024ff7","subject":"Merge \"[OpenStackSDK] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/326374236b856511450406508f5276595d024ff7"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:16.000000000","tz":600},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-19 03:42:04.000000000","tz":480},"subject":"[Puppet OpenStack] final releases for pike","message":"[Puppet OpenStack] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I7efab90fc9ebb7e4bc3f0e59f07c51a972414062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c2badcf773f2e6bfbfa1457bbf2c515949efcbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c2badcf773f2e6bfbfa1457bbf2c515949efcbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
