)]}'
{"id":"openstack%2Freleases~652915","triplet_id":"openstack%2Freleases~master~I9d2394ee3a2cfcf065bf54a1bd1d600ce2ccdf43","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I9d2394ee3a2cfcf065bf54a1bd1d600ce2ccdf43","subject":"[Puppet OpenStack] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 06:01:53.000000000","updated":"2019-05-23 06:25:17.000000000","submitted":"2019-05-23 06:25:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652915-1558592717943-13ee6a08","meta_rev_id":"dc5fc3a973a9d01702dbd6f574123eba85aeac9a","_number":652915,"virtual_id_number":652915,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-23 06:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-05-23 05:57:11.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-05-20 02:44:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-05-23 05:57:11.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 02:44:30.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-05-23 06:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddba29d8e61d110256db91a587435428cf88f749","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 06:01:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a35afce5f51870a5f5c2f84bfc1d4fc63ba11ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 09:20:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/652915/1/check/openstack-tox-docs/a671ec3/html/ : SUCCESS in 7m 10s\n- openstack-tox-validate http://logs.openstack.org/15/652915/1/check/openstack-tox-validate/4c1f9de/ : FAILURE in 4m 46s\n- releases-tox-list-changes http://logs.openstack.org/15/652915/1/check/releases-tox-list-changes/6d84c26/ : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9b99f76e522e6c868257712438ac57254edda948","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-17 03:10:03.000000000","message":"Patch Set 1:\n\nWaiting for these patches[1] to merge. Those hash values should be changed. All puppet module versions are unified to 11.6.0, except puppet-vswitch（7.5.0）, puppet-vitrage（1.2.0）\n[1]https://review.openstack.org/#/q/topic:pre-release-pike-final+(status:open+OR+status:merged)","accounts_in_message":[],"_revision_number":1},{"id":"3b5b6e5f59ead2b119f82f541731ecb5e864ec87","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-17 14:30:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2b025d3365e32e8b2a812ce6c79f1fb1473955c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 14:42:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/652915/2/check/openstack-tox-docs/493ac5b/html/ : SUCCESS in 7m 18s\n- openstack-tox-validate http://logs.openstack.org/15/652915/2/check/openstack-tox-validate/e194f8d/ : SUCCESS in 5m 35s\n- releases-tox-list-changes http://logs.openstack.org/15/652915/2/check/releases-tox-list-changes/59485bb/ : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":2},{"id":"fb975cc38635c0c502b9e32236985f8eeab3dc1d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 02:44:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8da024ee201919ab6bf04674ec227a294ae71345","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 05:57:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nWe discussed this in several team meetings.  This is the last pike-em change.  I know it\u0027s pretty close to self approving but it\u0027s pretty safe and we\u0027re down several team members this week","accounts_in_message":[],"_revision_number":2},{"id":"1cc22792cce1a7fdde05b33dd24423028048515c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 05:57:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"15086d20b8077ae852c71084a939b324dc1584ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:25:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"61448fb93a79b88aa9a4859bc20e2ca2255b373f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:25:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/652915/2/gate/openstack-tox-docs/de80fc2/html/ : SUCCESS in 6m 34s\n- openstack-tox-validate http://logs.openstack.org/15/652915/2/gate/openstack-tox-validate/260e0d5/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a35fac636c6c686c0c9c4a47b5dd45855ad32bb2","revisions":{"45b54d9393be621bf63d9d7877c8efb9ebbe13a4":{"kind":"REWORK","_number":1,"created":"2019-04-16 06:01:53.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/15/652915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/652915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/652915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/652915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/652915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/652915/1"}}},"commit":{"parents":[{"commit":"9b1f117c5f07d88623cccea8f0c079a0e8fe286a","subject":"[Puppet OpenStack] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b1f117c5f07d88623cccea8f0c079a0e8fe286a"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"subject":"[Puppet OpenStack] create pike-em tag against final releases","message":"[Puppet OpenStack] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I9d2394ee3a2cfcf065bf54a1bd1d600ce2ccdf43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45b54d9393be621bf63d9d7877c8efb9ebbe13a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45b54d9393be621bf63d9d7877c8efb9ebbe13a4"}]},"branch":"refs/heads/master"},"a35fac636c6c686c0c9c4a47b5dd45855ad32bb2":{"kind":"REWORK","_number":2,"created":"2019-05-17 14:30:13.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/15/652915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/652915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/652915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/652915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/652915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/652915/2"}}},"commit":{"parents":[{"commit":"07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc","subject":"Merge \"[tricircle] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:19.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-17 14:27:18.000000000","tz":-300},"subject":"[Puppet OpenStack] create pike-em tag against final releases","message":"[Puppet OpenStack] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I9d2394ee3a2cfcf065bf54a1bd1d600ce2ccdf43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a35fac636c6c686c0c9c4a47b5dd45855ad32bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a35fac636c6c686c0c9c4a47b5dd45855ad32bb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
