)]}'
{"id":"openstack%2Freleases~653181","triplet_id":"openstack%2Freleases~master~I6a6cf9110a737204e332c60004f72a68b770d1e7","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I6a6cf9110a737204e332c60004f72a68b770d1e7","subject":"[list_changes] Handle non-numeric tags","status":"MERGED","created":"2019-04-17 01:22:43.000000000","updated":"2019-05-07 19:10:55.000000000","submitted":"2019-05-07 19:10:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653181-1557256255144-85cee9de","meta_rev_id":"e32b39edda71dc8aa1c4840009c84934e4af900c","_number":653181,"virtual_id_number":653181,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-05-07 19:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-07 18:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-18 08:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-07 18:58:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 08:57:37.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-07 18:58:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-07 19:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"829e97980e549da63e607053a40c50f2495cf2d7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 01:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"223fb6b62d37a2db99f01f84f6c47e61c0e91400","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 01:31:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/81/653181/1/check/openstack-tox-py35/fa7a37d/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/81/653181/1/check/openstack-tox-py36/1d5830d/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/81/653181/1/check/openstack-tox-docs/063d153/html/ : SUCCESS in 6m 13s\n- openstack-tox-validate http://logs.openstack.org/81/653181/1/check/openstack-tox-validate/66b2fdf/ : SUCCESS in 5m 14s\n- releases-tox-list-changes http://logs.openstack.org/81/653181/1/check/releases-tox-list-changes/d2d0f42/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/81/653181/1/check/openstack-tox-pep8/f419383/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8656c69628f05c4e0eedc77a60b4c31cb349c189","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-18 08:55:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"05c109ec50ed1688ee6c48a3248815e186732e56","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-18 08:57:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35ef43e997ba523565cf6faf548195665adffe11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 09:21:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/81/653181/2/check/openstack-tox-py35/3e6aacc/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/81/653181/2/check/openstack-tox-py36/24f8f34/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/81/653181/2/check/openstack-tox-docs/cd2079d/html/ : SUCCESS in 8m 30s\n- openstack-tox-validate http://logs.openstack.org/81/653181/2/check/openstack-tox-validate/efb5a30/ : SUCCESS in 8m 18s\n- releases-tox-list-changes http://logs.openstack.org/81/653181/2/check/releases-tox-list-changes/885428c/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/81/653181/2/check/openstack-tox-pep8/2c4d034/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"aebfa1ba6f8e72bfabac1381b79c77fb4c342715","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-07 18:58:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"16ad64dc56f5fb84b409bb6cb96bc92293d72330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 18:58:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6bff5ce81d4bb33a7dbfd204a859a6d709e7ce63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 19:10:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/81/653181/2/gate/openstack-tox-py35/401cd11/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/81/653181/2/gate/openstack-tox-py36/41fe9e7/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/81/653181/2/gate/openstack-tox-docs/619c482/html/ : SUCCESS in 6m 32s\n- openstack-tox-validate http://logs.openstack.org/81/653181/2/gate/openstack-tox-validate/f9ba7f2/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e32b39edda71dc8aa1c4840009c84934e4af900c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 19:10:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55fa425cdc0617da2ae94de37c4452dc92808c79","revisions":{"a23dea7dfdbf08890a06af6fb624491c523e8faa":{"kind":"REWORK","_number":1,"created":"2019-04-17 01:22:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/81/653181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/653181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/653181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/653181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/653181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/653181/1"}}},"commit":{"parents":[{"commit":"78afdabc0c2dab74a0f513a7cb7c5777d211854e","subject":"Merge \"[octavia] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78afdabc0c2dab74a0f513a7cb7c5777d211854e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 01:15:34.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 01:22:07.000000000","tz":600},"subject":"[list_chnages] Handle non-numeri tags","message":"[list_chnages] Handle non-numeri tags\n\nAfter we tagged a realease EM git describe on a branch will include that\ntag and as it contains a \u0027-\u0027 it breaks the partition() we\u0027re using to\nget the tag[1].\n\nSwitch to using a re rather than the simple partition.\n\n(venv) [tony@thor releases]$ ipython\nPython 3.7.3 (default, Mar 27 2019, 13:41:07)\nType \u0027copyright\u0027, \u0027credits\u0027 or \u0027license\u0027 for more information\nIPython 7.4.0 -- An enhanced Interactive Python. Type \u0027?\u0027 for help.\n\nIn [1]: import openstack_releases.cmds.list_changes\nIn [2]: workdir \u003d \u0027/home/tony/projects/openstack\u0027\nIn [3]: repo \u003d \u0027openstack/python-manilaclient\u0027\nIn [4]: openstack_releases.cmds.list_changes.git_list_existing_branches(workdir, repo)\n\nAll Branches with Version Numbers\n---------------------------------\nfeature/add-constraints-support 1.11.0-36-g7134e3c   1.11.0       2 years, 8 months ago\nmaster                         1.27.0-10-g05a3f4d   1.27.0       6 months ago\nreview/openstack_release_bot/create-stein 1.27.0-2-g31608cf    1.27.0       6 months ago\nremotes/gerrit/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nfatal: ambiguous argument \u0027ocata\u0027: unknown revision or path not in the working tree.\nUse \u0027--\u0027 to separate paths from revisions, like this:\n\u0027git \u003ccommand\u003e [\u003crevision\u003e...] -- [\u003cfile\u003e...]\u0027\nremotes/gerrit/stable/ocata    ocata-em-5-g22e9c42  ocata\nremotes/gerrit/stable/pike     1.17.4               1.17.4       7 months ago\nremotes/gerrit/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/gerrit/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/gerrit/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\nremotes/origin/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nfatal: ambiguous argument \u0027ocata\u0027: unknown revision or path not in the working tree.\nUse \u0027--\u0027 to separate paths from revisions, like this:\n\u0027git \u003ccommand\u003e [\u003crevision\u003e...] -- [\u003cfile\u003e...]\u0027\nremotes/origin/stable/ocata    ocata-em-5-g22e9c42  ocata\nremotes/origin/stable/pike     1.17.4               1.17.4       7 months ago\nremotes/origin/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/origin/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/origin/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\n\nIn [5]: from imp import reload\nIn [6]: reload(openstack_releases.cmds.list_changes)\nOut[6]: \u003cmodule \u0027openstack_releases.cmds.list_changes\u0027 from \u0027/home/tony/projects/openstack/openstack/releases/openstack_releases/cmds/list_changes.py\u0027\u003e\nIn [7]: openstack_releases.cmds.list_changes.git_list_existing_branches(workdir, repo)\n\nAll Branches with Version Numbers\n---------------------------------\nfeature/add-constraints-support 1.11.0-36-g7134e3c   1.11.0       2 years, 8 months ago\nmaster                         1.27.0-10-g05a3f4d   1.27.0       6 months ago\nreview/openstack_release_bot/create-stein 1.27.0-2-g31608cf    1.27.0       6 months ago\nremotes/gerrit/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nremotes/gerrit/stable/ocata    ocata-em-5-g22e9c42  ocata-em     2 years, 3 months ago\nremotes/gerrit/stable/pike     1.17.4\nremotes/gerrit/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/gerrit/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/gerrit/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\nremotes/origin/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nremotes/origin/stable/ocata    ocata-em-5-g22e9c42  ocata-em     2 years, 3 months ago\nremotes/origin/stable/pike     1.17.4\nremotes/origin/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/origin/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/origin/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\n\n[1] http://logs.openstack.org/75/652775/2/check/releases-tox-list-changes/92b219a/job-output.txt.gz#_2019-04-16_14_25_50_377843\n\nChange-Id: I6a6cf9110a737204e332c60004f72a68b770d1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a23dea7dfdbf08890a06af6fb624491c523e8faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a23dea7dfdbf08890a06af6fb624491c523e8faa"}]},"branch":"refs/heads/master"},"55fa425cdc0617da2ae94de37c4452dc92808c79":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-18 08:55:42.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/81/653181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/653181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/653181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/653181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/653181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/653181/2"}}},"commit":{"parents":[{"commit":"78afdabc0c2dab74a0f513a7cb7c5777d211854e","subject":"Merge \"[octavia] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78afdabc0c2dab74a0f513a7cb7c5777d211854e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 01:15:34.000000000","tz":600},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-04-18 08:55:33.000000000","tz":0},"subject":"[list_changes] Handle non-numeric tags","message":"[list_changes] Handle non-numeric tags\n\nAfter we tagged a realease EM git describe on a branch will include that\ntag and as it contains a \u0027-\u0027 it breaks the partition() we\u0027re using to\nget the tag[1].\n\nSwitch to using a re rather than the simple partition.\n\n(venv) [tony@thor releases]$ ipython\nPython 3.7.3 (default, Mar 27 2019, 13:41:07)\nType \u0027copyright\u0027, \u0027credits\u0027 or \u0027license\u0027 for more information\nIPython 7.4.0 -- An enhanced Interactive Python. Type \u0027?\u0027 for help.\n\nIn [1]: import openstack_releases.cmds.list_changes\nIn [2]: workdir \u003d \u0027/home/tony/projects/openstack\u0027\nIn [3]: repo \u003d \u0027openstack/python-manilaclient\u0027\nIn [4]: openstack_releases.cmds.list_changes.git_list_existing_branches(workdir, repo)\n\nAll Branches with Version Numbers\n---------------------------------\nfeature/add-constraints-support 1.11.0-36-g7134e3c   1.11.0       2 years, 8 months ago\nmaster                         1.27.0-10-g05a3f4d   1.27.0       6 months ago\nreview/openstack_release_bot/create-stein 1.27.0-2-g31608cf    1.27.0       6 months ago\nremotes/gerrit/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nfatal: ambiguous argument \u0027ocata\u0027: unknown revision or path not in the working tree.\nUse \u0027--\u0027 to separate paths from revisions, like this:\n\u0027git \u003ccommand\u003e [\u003crevision\u003e...] -- [\u003cfile\u003e...]\u0027\nremotes/gerrit/stable/ocata    ocata-em-5-g22e9c42  ocata\nremotes/gerrit/stable/pike     1.17.4               1.17.4       7 months ago\nremotes/gerrit/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/gerrit/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/gerrit/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\nremotes/origin/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nfatal: ambiguous argument \u0027ocata\u0027: unknown revision or path not in the working tree.\nUse \u0027--\u0027 to separate paths from revisions, like this:\n\u0027git \u003ccommand\u003e [\u003crevision\u003e...] -- [\u003cfile\u003e...]\u0027\nremotes/origin/stable/ocata    ocata-em-5-g22e9c42  ocata\nremotes/origin/stable/pike     1.17.4               1.17.4       7 months ago\nremotes/origin/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/origin/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/origin/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\n\nIn [5]: from imp import reload\nIn [6]: reload(openstack_releases.cmds.list_changes)\nOut[6]: \u003cmodule \u0027openstack_releases.cmds.list_changes\u0027 from \u0027/home/tony/projects/openstack/openstack/releases/openstack_releases/cmds/list_changes.py\u0027\u003e\nIn [7]: openstack_releases.cmds.list_changes.git_list_existing_branches(workdir, repo)\n\nAll Branches with Version Numbers\n---------------------------------\nfeature/add-constraints-support 1.11.0-36-g7134e3c   1.11.0       2 years, 8 months ago\nmaster                         1.27.0-10-g05a3f4d   1.27.0       6 months ago\nreview/openstack_release_bot/create-stein 1.27.0-2-g31608cf    1.27.0       6 months ago\nremotes/gerrit/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nremotes/gerrit/stable/ocata    ocata-em-5-g22e9c42  ocata-em     2 years, 3 months ago\nremotes/gerrit/stable/pike     1.17.4\nremotes/gerrit/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/gerrit/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/gerrit/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\nremotes/origin/master          1.27.0-10-g05a3f4d   1.27.0       6 months ago\nremotes/origin/stable/ocata    ocata-em-5-g22e9c42  ocata-em     2 years, 3 months ago\nremotes/origin/stable/pike     1.17.4\nremotes/origin/stable/queens   1.21.1-6-g02d6234    1.21.1       1 year, 1 month ago\nremotes/origin/stable/rocky    1.24.1-9-g816bfc8    1.24.1       9 months ago\nremotes/origin/stable/stein    1.27.0-3-g7430f4c    1.27.0       6 months ago\n\n[1] http://logs.openstack.org/75/652775/2/check/releases-tox-list-changes/92b219a/job-output.txt.gz#_2019-04-16_14_25_50_377843\n\nChange-Id: I6a6cf9110a737204e332c60004f72a68b770d1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/55fa425cdc0617da2ae94de37c4452dc92808c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/55fa425cdc0617da2ae94de37c4452dc92808c79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
