)]}'
{"id":"openstack%2Freleases~653689","triplet_id":"openstack%2Freleases~master~I768f7c1fcd43fb605a714cf4f1da2943d10866ad","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I768f7c1fcd43fb605a714cf4f1da2943d10866ad","subject":"Release python-monascaclient 1.15.0","status":"MERGED","created":"2019-04-18 08:32:31.000000000","updated":"2019-05-09 14:01:48.000000000","submitted":"2019-05-09 14:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653689-1557410508169-1a85bf1c","meta_rev_id":"612432f658e4ad0e271f1345d811f233269ba9c5","_number":653689,"virtual_id_number":653689,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"value":0,"date":"2019-04-23 10:58:57.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-05-09 14:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-23 12:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":2,"date":"2019-05-09 13:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-23 12:54:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-04-18 08:33:16.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":1,"date":"2019-05-09 13:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-04-18 08:33:10.000000000","_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 08:33:10.000000000","updated_by":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"reviewer":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"state":"REVIEWER"},{"updated":"2019-04-18 08:33:16.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-04-23 12:07:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-23 12:54:42.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-05-09 13:49:35.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-09 14:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ddbe4d907986337d319ccd8998516c04fdcddb1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-18 08:32:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5c001f47895c0a46387356b7a4e3c7a7e2408d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 08:53:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/653689/1/check/openstack-tox-docs/6dd6716/html/ : SUCCESS in 7m 10s\n- openstack-tox-validate http://logs.openstack.org/89/653689/1/check/openstack-tox-validate/bd5ea13/ : FAILURE in 4m 45s\n- releases-tox-list-changes http://logs.openstack.org/89/653689/1/check/releases-tox-list-changes/b098ca8/ : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"391ed94167aace4c10a0f25ab59d14cf65b53eda","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-18 09:25:07.000000000","message":"Patch Set 1:\n\nThis will fail tests until https://review.openstack.org/#/c/653685/ lands","accounts_in_message":[],"_revision_number":1},{"id":"98c1fa5876fa2a1cd21e316b5a408ae4575c3113","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 18:42:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11e206bd44662764a2bd1a26229ab41d883f2393","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:54:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/653689/1/check/openstack-tox-docs/0e0679c/html/ : SUCCESS in 5m 58s\n- openstack-tox-validate http://logs.openstack.org/89/653689/1/check/openstack-tox-validate/9480943/ : SUCCESS in 4m 21s\n- releases-tox-list-changes http://logs.openstack.org/89/653689/1/check/releases-tox-list-changes/fca98a2/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c348a5c3f13fe61221c3af5aac0b19208806e4d5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 19:33:56.000000000","message":"Patch Set 1:\n\nI only see the bugfix. Is there a reason for the minor version bump instead of making this 1.14.1?","accounts_in_message":[],"_revision_number":1},{"id":"03ea1a82e2ad486ae3ea19b530d48c806b0f6213","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-23 08:17:16.000000000","message":"Patch Set 1:\n\nIn addition to the single Python bugfix, a Dockerfile has been added (see: 112f8c0402a4d6d5d7c32da23c34f30447ec3f3b).","accounts_in_message":[],"_revision_number":1},{"id":"b1848cd29129e28dd7bbc897c1f8664ad0897129","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-23 10:07:44.000000000","message":"Patch Set 1:\n\nI\u0027m not sure the Docker file qualifies as a \"feature\" that would justify the version bump... Thoughts?","accounts_in_message":[],"_revision_number":1},{"id":"c8409bf2f98c22b7ddfc936b4146ed1fad0ce4db","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-23 10:58:57.000000000","message":"Patch Set 1:\n\nThe Dockerfile is for users as well as for infra, and is documented in the README. For that reason, from a user perspective I see it as a feature, although I agree it isn\u0027t clear-cut. It\u0027s no problem for me if you would rather this was 1.14.1.","accounts_in_message":[],"_revision_number":1},{"id":"50fed175e779b19aae2582bdb5b3c37549494a6c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-23 12:07:40.000000000","message":"Patch Set 1: Code-Review+2\n\nAh, that makes sense. Thanks for the explanation.","accounts_in_message":[],"_revision_number":1},{"id":"eb0f2854c06a27d5955b09c7c010f4936b3e823a","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-04-23 12:54:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fccc6d91947f0a6fbb628b91f392f9ecfced7b69","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-09 13:49:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ff7ca6186e7c1916d14a9621c212cfc03e72957","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:49:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe40c569e89349c08b7ed645d3f13e6a32bb0dc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:01:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/653689/1/gate/openstack-tox-docs/bc3271d/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/89/653689/1/gate/openstack-tox-validate/3090dc9/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"612432f658e4ad0e271f1345d811f233269ba9c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:01:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dae89d680971004dca0a898daa8566a58c2f97fb","revisions":{"dae89d680971004dca0a898daa8566a58c2f97fb":{"kind":"REWORK","_number":1,"created":"2019-04-18 08:32:31.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/89/653689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/653689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/653689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/653689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/653689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/653689/1"}}},"commit":{"parents":[{"commit":"fe4d9d87b3bb841a1b75c9bcbfbbc144eee3744e","subject":"Merge \"[keystone] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe4d9d87b3bb841a1b75c9bcbfbbc144eee3744e"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-18 08:23:08.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-18 08:23:08.000000000","tz":60},"subject":"Release python-monascaclient 1.15.0","message":"Release python-monascaclient 1.15.0\n\nThis includes a bugfix to correct formatting under Python3.\n\nDepends-On: https://review.openstack.org/#/c/653685/\nChange-Id: I768f7c1fcd43fb605a714cf4f1da2943d10866ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dae89d680971004dca0a898daa8566a58c2f97fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dae89d680971004dca0a898daa8566a58c2f97fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
