)]}'
{"id":"openstack%2Freleases~653793","triplet_id":"openstack%2Freleases~master~I5dd46957383749290872504d0648d29198e35407","project":"openstack/releases","branch":"master","topic":"oslo-ff-train","hashtags":[],"change_id":"I5dd46957383749290872504d0648d29198e35407","subject":"Add Oslo feature freeze to Train schedule","status":"MERGED","created":"2019-04-18 15:23:33.000000000","updated":"2019-04-22 21:52:45.000000000","submitted":"2019-04-22 21:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653793-1555969965576-a663f06a","meta_rev_id":"9bb102a8b07d39cf64800ecc3375fcd338da103d","_number":653793,"virtual_id_number":653793,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-22 21:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-04-22 21:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-04-22 21:39:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 21:39:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-22 21:52:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b1617e9892484b7bf17556b1eea5215adee4d5c","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-18 15:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"004bef03c9077d44c0107f1487a80b2ede8c8828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 15:37:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/653793/1/check/openstack-tox-docs/c4decaa/html/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"03cae0c3a4bfc93b9250b5bc4df1b326328f3615","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-18 15:41:01.000000000","message":"Patch Set 1: Workflow-1\n\nProbably should have put this in the project-specific column.","accounts_in_message":[],"_revision_number":1},{"id":"1722139909c035d8145dc298c9feddf77ab7b351","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-18 15:42:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e7a50d29991fcd0251aae52859be43f2e7a855e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 15:54:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/653793/2/check/openstack-tox-docs/da1dea2/html/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a7a460756fbd992beb394f87b151e0b65bc92c11","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:39:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fd867a167e89cfb6cfab9d2e277b33aa445d0db5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:40:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b790eda10de4aa034688610b3f697cff6e0f993","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:52:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/653793/2/gate/openstack-tox-docs/1761786/html/ : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":2},{"id":"396555e1fcad99f7ece6ff1c2b279e90d2dfb346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:52:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2acb8a1e6f43fdc436cf4f32faf0e16969731a7d","revisions":{"2130d7b8b1d4572ffb7f77e1aa0fae26910f5b74":{"kind":"REWORK","_number":1,"created":"2019-04-18 15:23:33.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/93/653793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/653793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/653793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/653793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/653793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/653793/1"}}},"commit":{"parents":[{"commit":"a6e1edca0d07b1ee10e26bb79b16c4d757d3cd40","subject":"Add Train cycle schedule for keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6e1edca0d07b1ee10e26bb79b16c4d757d3cd40"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-18 15:22:39.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-18 15:22:39.000000000","tz":0},"subject":"Add Oslo feature freeze to Train schedule","message":"Add Oslo feature freeze to Train schedule\n\nPer [1], Oslo observes feature freeze earlier than most projects in\norder to give consuming projects a chance to implement any new\nfeatures in Oslo before their feature freeze.  This should be\nreflected in the release schedule.\n\nChange-Id: I5dd46957383749290872504d0648d29198e35407\n1: http://specs.openstack.org/openstack/oslo-specs/specs/policy/feature-freeze.html\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2130d7b8b1d4572ffb7f77e1aa0fae26910f5b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2130d7b8b1d4572ffb7f77e1aa0fae26910f5b74"}]},"branch":"refs/heads/master"},"2acb8a1e6f43fdc436cf4f32faf0e16969731a7d":{"kind":"REWORK","_number":2,"created":"2019-04-18 15:42:28.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/93/653793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/653793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/653793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/653793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/653793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/653793/2"}}},"commit":{"parents":[{"commit":"a6e1edca0d07b1ee10e26bb79b16c4d757d3cd40","subject":"Add Train cycle schedule for keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a6e1edca0d07b1ee10e26bb79b16c4d757d3cd40"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-18 15:22:39.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-18 15:41:57.000000000","tz":0},"subject":"Add Oslo feature freeze to Train schedule","message":"Add Oslo feature freeze to Train schedule\n\nPer [1], Oslo observes feature freeze earlier than most projects in\norder to give consuming projects a chance to implement any new\nfeatures in Oslo before their feature freeze.  This should be\nreflected in the release schedule.\n\nChange-Id: I5dd46957383749290872504d0648d29198e35407\n1: http://specs.openstack.org/openstack/oslo-specs/specs/policy/feature-freeze.html\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2acb8a1e6f43fdc436cf4f32faf0e16969731a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2acb8a1e6f43fdc436cf4f32faf0e16969731a7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
