)]}'
{"id":"openstack%2Freleases~654620","triplet_id":"openstack%2Freleases~master~I3b9441931e04056a848bab7dbeb80b70a343f1db","project":"openstack/releases","branch":"master","topic":"oslo-master","hashtags":[],"change_id":"I3b9441931e04056a848bab7dbeb80b70a343f1db","subject":"Oslo releases for 2019-4-22","status":"MERGED","created":"2019-04-22 19:50:43.000000000","updated":"2019-04-26 16:40:31.000000000","submitted":"2019-04-26 16:40:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654620-1556296831088-47422c5d","meta_rev_id":"99ac42531884fe7d1e19894f73ed374a3f3339e9","_number":654620,"virtual_id_number":654620,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-04-23 10:01:54.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-04-26 16:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-22 21:38:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-04-25 13:51:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-04-26 16:27:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:01:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-04-25 13:51:22.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-04-26 16:27:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-26 16:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b13bfbbf2a5a6fbd88ac06815729fefa5da0c51","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-22 19:50:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be78a8ac340ca522e67f54f0cd790d4d860f595a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:04:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/20/654620/1/check/openstack-tox-docs/37d2de5/html/ : SUCCESS in 7m 19s\n- openstack-tox-validate http://logs.openstack.org/20/654620/1/check/openstack-tox-validate/5215338/ : FAILURE in 7m 06s\n- releases-tox-list-changes http://logs.openstack.org/20/654620/1/check/releases-tox-list-changes/05dee28/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"73c4053dbcfe497fcc9fa585fa54f50ff095184d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 20:27:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3fb58a61e3c4c25d751a91a1bbcc55c66197477","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:39:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/20/654620/2/check/openstack-tox-docs/0223005/html/ : SUCCESS in 7m 19s\n- openstack-tox-validate http://logs.openstack.org/20/654620/2/check/openstack-tox-validate/d35f758/ : SUCCESS in 7m 05s\n- releases-tox-list-changes http://logs.openstack.org/20/654620/2/check/releases-tox-list-changes/5e40cd0/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2c226139883ac84bbfeb764f0c07eb6520133df2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:38:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e7f40ea51dac99827e8172ed4c9c7d2fd83e71db","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-23 10:01:54.000000000","message":"Patch Set 2:\n\nLGTM but holding until we get a better feel of that twine issue that made yesterdays Pike releases all fail","accounts_in_message":[],"_revision_number":2},{"id":"ef1359b6a7676a38bb44436d391f1b3a906054f1","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-25 13:51:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe29ac93d9f7c962e7ac5dd80089617af4222c2a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-26 16:27:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15dbe704f2fa3b75a5bcdf44ff976c8b91964d72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:27:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df1e1d8a85e7c7a4720cfae8b08afc331deb7495","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:40:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/20/654620/2/gate/openstack-tox-docs/706673f/html/ : SUCCESS in 7m 27s\n- openstack-tox-validate http://logs.openstack.org/20/654620/2/gate/openstack-tox-validate/44ef248/ : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e8878e36b754ff734c3cae10df30ad7c47818acb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:40:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"136a049ed0e08c5c8970bd7413861fdbea49fbbc","revisions":{"52a36440e52149dbeffeddd68bc23a59137e8379":{"kind":"REWORK","_number":1,"created":"2019-04-22 19:50:43.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/20/654620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/654620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/654620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/654620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/654620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/654620/1"}}},"commit":{"parents":[{"commit":"e4695cc987ed86fbd40dd8566bb9620c3415209b","subject":"Merge \"Switch monasca-thresh to the new \u0027generic\u0027 release-type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4695cc987ed86fbd40dd8566bb9620c3415209b"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-22 19:50:25.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-22 19:50:25.000000000","tz":0},"subject":"Oslo releases for 2019-4-22","message":"Oslo releases for 2019-4-22\n\nChange-Id: I3b9441931e04056a848bab7dbeb80b70a343f1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52a36440e52149dbeffeddd68bc23a59137e8379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52a36440e52149dbeffeddd68bc23a59137e8379"}]},"branch":"refs/heads/master"},"136a049ed0e08c5c8970bd7413861fdbea49fbbc":{"kind":"REWORK","_number":2,"created":"2019-04-22 20:27:00.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/20/654620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/654620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/654620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/654620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/654620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/654620/2"}}},"commit":{"parents":[{"commit":"e4695cc987ed86fbd40dd8566bb9620c3415209b","subject":"Merge \"Switch monasca-thresh to the new \u0027generic\u0027 release-type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4695cc987ed86fbd40dd8566bb9620c3415209b"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-22 19:50:25.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:26:53.000000000","tz":-300},"subject":"Oslo releases for 2019-4-22","message":"Oslo releases for 2019-4-22\n\nChange-Id: I3b9441931e04056a848bab7dbeb80b70a343f1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/136a049ed0e08c5c8970bd7413861fdbea49fbbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/136a049ed0e08c5c8970bd7413861fdbea49fbbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
