)]}'
{"id":"openstack%2Freleases~654623","triplet_id":"openstack%2Freleases~master~I41e2422e9ac48f0b7df0ed4ac88060dbc7007439","project":"openstack/releases","branch":"master","topic":"oslo-rocky","hashtags":[],"change_id":"I41e2422e9ac48f0b7df0ed4ac88060dbc7007439","subject":"oslo.db 4.40.2","status":"MERGED","created":"2019-04-22 20:04:13.000000000","updated":"2019-05-07 19:14:48.000000000","submitted":"2019-05-07 19:14:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654623-1557256488662-ef73585b","meta_rev_id":"11a331c009b378c758e0451e352c90ae43db0beb","_number":654623,"virtual_id_number":654623,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-05-07 19:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-07 19:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-06 20:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-05-06 20:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-07 19:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:09:06.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-04-25 13:50:40.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-07 19:02:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-07 19:14:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4ecbba2a78a7cb90a58687f859a88bf2a66d4d5","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-22 20:04:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37efdecbd6793b24eb6f1268b8c707c618649570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:18:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/23/654623/1/check/openstack-tox-docs/20611c1/html/ : SUCCESS in 7m 03s\n- openstack-tox-validate http://logs.openstack.org/23/654623/1/check/openstack-tox-validate/baee971/ : SUCCESS in 5m 31s\n- releases-tox-list-changes http://logs.openstack.org/23/654623/1/check/releases-tox-list-changes/addda1b/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c413010a90dbbe6bc06a944c2cf0382839fd035e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-23 10:09:06.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"d84e3eb0a8fa8ffd51d768ea83aff6c89c07e27e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-25 13:50:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c0e516adf0c21734f3aa9ca8622761aeea21f638","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-06 20:33:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d0471cb3765dab80a43e2fad73f2e459c510a641","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 20:41:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/23/654623/2/check/openstack-tox-docs/85bb4ea/html/ : SUCCESS in 6m 26s\n- openstack-tox-validate http://logs.openstack.org/23/654623/2/check/openstack-tox-validate/2337d87/ : SUCCESS in 4m 00s\n- releases-tox-list-changes http://logs.openstack.org/23/654623/2/check/releases-tox-list-changes/455fa4b/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"90f7ff76514d5a85569f39ef472241c7d954cf4f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-07 19:02:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3d7ab674724e7be0cb4a1a18e031008d1010ebbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 19:02:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db8d9034acbf1007f433cf4e45cdfad58410411f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 19:14:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/23/654623/2/gate/openstack-tox-docs/7dcdfc7/html/ : SUCCESS in 6m 48s\n- openstack-tox-validate http://logs.openstack.org/23/654623/2/gate/openstack-tox-validate/639f4eb/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2ce92980b8911ad058a35f87aa82d123937347cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 19:14:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"659f9f359013cdcbfac16acfda0f23ac24b1dfdc","revisions":{"f7fae319f98a1683ec70cf3c8aa8e5f3d560a45b":{"kind":"REWORK","_number":1,"created":"2019-04-22 20:04:13.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/23/654623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/654623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/654623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/654623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/654623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/654623/1"}}},"commit":{"parents":[{"commit":"79fa94bf802cd066957fdedc8c4603144f842205","subject":"Merge \"[horizon] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79fa94bf802cd066957fdedc8c4603144f842205"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-22 20:03:50.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-22 20:03:50.000000000","tz":0},"subject":"oslo.db 4.40.2","message":"oslo.db 4.40.2\n\nThis is a bugfix release of stable/rocky.\n\nChange-Id: I41e2422e9ac48f0b7df0ed4ac88060dbc7007439\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7fae319f98a1683ec70cf3c8aa8e5f3d560a45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7fae319f98a1683ec70cf3c8aa8e5f3d560a45b"}]},"branch":"refs/heads/master"},"659f9f359013cdcbfac16acfda0f23ac24b1dfdc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-06 20:33:55.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/23/654623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/654623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/654623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/654623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/654623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/654623/2"}}},"commit":{"parents":[{"commit":"61b3779ff2bd0de730c6615d67eab8cd1d9de29a","subject":"Merge \"Release openstackdocstheme 1.29.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/61b3779ff2bd0de730c6615d67eab8cd1d9de29a"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-04-22 20:03:50.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-06 20:27:39.000000000","tz":0},"subject":"oslo.db 4.40.2","message":"oslo.db 4.40.2\n\nThis is a bugfix release of stable/rocky.\n\nChange-Id: I41e2422e9ac48f0b7df0ed4ac88060dbc7007439\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/659f9f359013cdcbfac16acfda0f23ac24b1dfdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/659f9f359013cdcbfac16acfda0f23ac24b1dfdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
