)]}'
{"id":"openstack%2Freleases~654627","triplet_id":"openstack%2Freleases~master~I12d8738da634a76da693cc15db07aac176f08958","project":"openstack/releases","branch":"master","topic":"migration","hashtags":[],"change_id":"I12d8738da634a76da693cc15db07aac176f08958","subject":"Move openstack-infra to new namespaces","status":"MERGED","created":"2019-04-22 20:42:14.000000000","updated":"2019-06-13 10:03:18.000000000","submitted":"2019-06-13 10:03:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654627-1560420198898-3fe75134","meta_rev_id":"352ed4ed1df7abb86c73e0870794974a12f769f2","_number":654627,"virtual_id_number":654627,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-06-13 10:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"value":2,"date":"2019-05-29 00:31:47.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-06-13 09:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-31 11:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-06-13 09:49:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 21:59:13.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-04-28 22:02:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-12 10:29:32.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-05-14 21:12:37.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-05-21 09:44:25.000000000","updated_by":{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},"reviewer":{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},"state":"REVIEWER"},{"updated":"2019-05-24 18:09:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-05-29 00:31:47.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-31 11:08:37.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-13 09:49:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-13 10:03:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7362f04a7fa09361bdb613e786fc5e561cc5fd1e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 20:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfa3f4d36f3ff559a680046014c05f99a4940afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:02:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/1/check/openstack-tox-docs/5a161dd/html/ : SUCCESS in 7m 17s\n- openstack-tox-validate http://logs.openstack.org/27/654627/1/check/openstack-tox-validate/d849107/ : FAILURE in 3m 54s\n- releases-tox-list-changes http://logs.openstack.org/27/654627/1/check/releases-tox-list-changes/85512de/ : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5b1e4b81445ea9d71dbfe55fd259420320bec67a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:27:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66b220ba7146dfb3329f8245adcba00a9ca97c1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:42:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/2/check/openstack-tox-docs/177e966/html/ : SUCCESS in 7m 04s\n- openstack-tox-validate http://logs.openstack.org/27/654627/2/check/openstack-tox-validate/a170ad0/ : FAILURE in 3m 39s\n- releases-tox-list-changes http://logs.openstack.org/27/654627/2/check/releases-tox-list-changes/2d79827/ : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"42bba3b38da2040d5cf4caae7c99ceb97f8c4048","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-23 13:32:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"195e08fa9a331e26fbad92c145a88623de27ec2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:56:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/3/check/openstack-tox-docs/96828f9/html/ : SUCCESS in 7m 31s\n- openstack-tox-validate http://logs.openstack.org/27/654627/3/check/openstack-tox-validate/08ddca0/ : SUCCESS in 6m 45s\n- releases-tox-list-changes http://logs.openstack.org/27/654627/3/check/releases-tox-list-changes/27949d4/ : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":3},{"id":"acc63f81bff5674e43072344bf8d81cc3a2559ce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-23 21:28:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"78e1b97ff01bd5b0d95f6a7d10dc6ae4a0b276a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 21:45:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/3/check/openstack-tox-docs/7f44018/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/27/654627/3/check/openstack-tox-validate/2385e6a/ : SUCCESS in 4m 39s\n- releases-tox-list-changes http://logs.openstack.org/27/654627/3/check/releases-tox-list-changes/b82495f/ : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"a9a4b158b6b70766833f4887f77a3489cf4527f9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-24 13:24:12.000000000","message":"Patch Set 3:\n\nI remember seeing a comment somewhere that opendev.org may not be updating all refs, which might explain why we don\u0027t see the tag and try to create it, then fail because it already exists. If that\u0027s the case, we may have some other issues.","accounts_in_message":[],"_revision_number":3},{"id":"3c8c3491dafdc4c54496902c18c784373a5f8da2","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-04-28 21:59:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a3df94d7a11df14c7b7708057ad6bb3f46efebf6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-28 22:02:32.000000000","message":"Patch Set 3:\n\nGood questions for some of the infra folks.","accounts_in_message":[],"_revision_number":3},{"id":"5747baf3e7b42434f671978ce3fa990bd058ac11","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 15:50:27.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cd67b6a104d9682888044f7eacebb25004528292","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-12 10:29:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c83b615e60df35e50b0c47732e4ecd83e8e18885","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-05-14 21:12:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ce8d1e6e8231ee59a87a1a4b0b02f441ea76a2d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-15 09:23:38.000000000","message":"Patch Set 3:\n\nWaiting for Kendall\u0027s input before further reviewing.\n\nnit: Should we point to governance changes for this?\nIt would clarify things of \"who owns what\".","accounts_in_message":[],"_revision_number":3},{"id":"c9dc47a08a444da38b17a759563c3dc5f1719c59","author":{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},"date":"2019-05-21 09:44:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d896d006f73dae9d842e2fb9eb1f1ee6c197cee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-21 16:55:51.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"20362f5f1348c3504fa5aeb032837c570289da60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:05:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/4/check/openstack-tox-docs/c7df88c/html/ : SUCCESS in 6m 38s\n- openstack-tox-validate http://logs.openstack.org/27/654627/4/check/openstack-tox-validate/4c7e29a/ : SUCCESS in 5m 05s\n- releases-tox-list-changes http://logs.openstack.org/27/654627/4/check/releases-tox-list-changes/a46412c/ : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"66a200fc2511e3e498fc7b86da3837c79e03549e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 00:19:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"62c50bb1588f2111cee648615cfab49fc46c10b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 00:40:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/5/check/openstack-tox-docs/ebb39c2/html/ : SUCCESS in 7m 15s\n- openstack-tox-validate http://logs.openstack.org/27/654627/5/check/openstack-tox-validate/52a3728/ : SUCCESS in 6m 13s\n- releases-tox-list-changes http://logs.openstack.org/27/654627/5/check/releases-tox-list-changes/0316d60/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":5},{"id":"ff961343dd4aecba5ca22aeb7f60ef5757a48241","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-24 10:11:44.000000000","message":"Patch Set 5: Code-Review-1\n\nSlight preference for consistency and having only openstack/ releases handled by the openstack release team and the openstack/releases repository. Happy to be overruled though :)","accounts_in_message":[],"_revision_number":5},{"id":"b2dc953e9203c664e6a855929e4f6a0a36bbe311","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-24 11:00:55.000000000","message":"Patch Set 5:\n\nDo we want to remove those now? I had considered the question of whether releases would now be done via the release team separate from this.","accounts_in_message":[],"_revision_number":5},{"id":"319911882a048a3218f280ea9223c61e184ab4b5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-24 18:09:17.000000000","message":"Patch Set 5:\n\nI asked the other StoryBoard maintainers for details on this, and apparently there was some discussion at the PTG about wanting to start doing more releases.\n\nDue to the lack of need for release coordination with the rest of OpenStack however, and the fact that releases can generally be performed easily enough by just pushing a signed tag to Gerrit, I\u0027m happy to help make opendev/python-storyboardclient (and I guess opendev/yaml2ical as well?) releases on request or assist with getting the necessary Gerrit ACLs in place to allow other developers on those projects do the same.","accounts_in_message":[],"_revision_number":5},{"id":"ccbc068e78fa70fc479f2f481d3cd9286138e01e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-28 02:09:54.000000000","message":"Patch Set 5:\n\n\u003e I asked the other StoryBoard maintainers for details on this, and\n \u003e apparently there was some discussion at the PTG about wanting to\n \u003e start doing more releases.\n \u003e \n \u003e Due to the lack of need for release coordination with the rest of\n \u003e OpenStack however, and the fact that releases can generally be\n \u003e performed easily enough by just pushing a signed tag to Gerrit, I\u0027m\n \u003e happy to help make opendev/python-storyboardclient (and I guess\n \u003e opendev/yaml2ical as well?) releases on request or assist with\n \u003e getting the necessary Gerrit ACLs in place to allow other\n \u003e developers on those projects do the same.\n\nOkay with that in mind I think we can drop those deliverables from this repo and followup with the *storyboard/yaml2ical releases on email/IRC.","accounts_in_message":[],"_revision_number":5},{"id":"cf8cdc8a26920b1c3cba9fd0544af9106148ff9b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-28 14:53:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ac007bb14194c54bb9a673dfb066241e28f0f417","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:12:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/6/check/openstack-tox-docs/a6b64f6/html/ : SUCCESS in 6m 47s\n- openstack-tox-validate http://logs.openstack.org/27/654627/6/check/openstack-tox-validate/aaca226/ : SUCCESS in 4m 09s\n- releases-tox-list-changes http://logs.openstack.org/27/654627/6/check/releases-tox-list-changes/1ed1ade/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":6},{"id":"ffab98050bca798b2a4d742ab36dcfe696fd9906","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-29 00:31:47.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bf027ee64a5ecc34d0db3e435a95cf8d9060d828","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-31 11:08:37.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9c861b82750d8ec9483f1dc3921a4ac968493355","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-13 09:49:43.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"bb47924d96bd9bc5fd56301cb0b6fdec8e4935a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:49:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"aa6e2e5a6364ead87d1ba9e06b5c7fea810c5388","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:03:18.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/654627/6/gate/openstack-tox-docs/773bad6/html/ : SUCCESS in 7m 55s\n- openstack-tox-validate http://logs.openstack.org/27/654627/6/gate/openstack-tox-validate/4f897df/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":6},{"id":"7e5253605d3a7e8ae6939e95970a4a429d242f2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:03:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"da5bc141bcacff868e1402d74e782b947974728c","revisions":{"f2b60be9fac80a7fe8a4cfc7ffd41ebdb7a3ca10":{"kind":"REWORK","_number":1,"created":"2019-04-22 20:42:14.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/27/654627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/654627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/654627/1"}}},"commit":{"parents":[{"commit":"6dc2c3445492243f6b344041d338575594515724","subject":"Merge \"stein/puppet-tripleo: release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dc2c3445492243f6b344041d338575594515724"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:42:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:42:08.000000000","tz":-300},"subject":"Move openstack-infra to openstack namespace","message":"Move openstack-infra to openstack namespace\n\nChange-Id: I12d8738da634a76da693cc15db07aac176f08958\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2b60be9fac80a7fe8a4cfc7ffd41ebdb7a3ca10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2b60be9fac80a7fe8a4cfc7ffd41ebdb7a3ca10"}]},"branch":"refs/heads/master"},"d5f0c6e08ab6055453b9cf9614f6ca1a45b76689":{"kind":"REWORK","_number":2,"created":"2019-04-22 21:27:52.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/27/654627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/654627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/654627/2"}}},"commit":{"parents":[{"commit":"6dc2c3445492243f6b344041d338575594515724","subject":"Merge \"stein/puppet-tripleo: release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dc2c3445492243f6b344041d338575594515724"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:42:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 21:27:14.000000000","tz":-300},"subject":"Move openstack-infra to openstack namespace","message":"Move openstack-infra to openstack namespace\n\nChange-Id: I12d8738da634a76da693cc15db07aac176f08958\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5f0c6e08ab6055453b9cf9614f6ca1a45b76689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5f0c6e08ab6055453b9cf9614f6ca1a45b76689"}]},"branch":"refs/heads/master"},"034761052e36491778aedf5dc095f98e73a584b7":{"kind":"REWORK","_number":3,"created":"2019-04-23 13:32:35.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/27/654627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/654627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/654627/3"}}},"commit":{"parents":[{"commit":"6dc2c3445492243f6b344041d338575594515724","subject":"Merge \"stein/puppet-tripleo: release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dc2c3445492243f6b344041d338575594515724"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:42:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-23 13:31:01.000000000","tz":-300},"subject":"Move openstack-infra to openstack namespace","message":"Move openstack-infra to openstack namespace\n\nChange-Id: I12d8738da634a76da693cc15db07aac176f08958\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/034761052e36491778aedf5dc095f98e73a584b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/034761052e36491778aedf5dc095f98e73a584b7"}]},"branch":"refs/heads/master"},"6e1cf88de8c3c32f387615a27dc00b56472d3236":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-05-21 16:55:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/27/654627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/654627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/654627/4"}}},"commit":{"parents":[{"commit":"6dc2c3445492243f6b344041d338575594515724","subject":"Merge \"stein/puppet-tripleo: release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dc2c3445492243f6b344041d338575594515724"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:42:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-21 16:55:36.000000000","tz":-300},"subject":"Move openstack-infra to new namespaces","message":"Move openstack-infra to new namespaces\n\nChange-Id: I12d8738da634a76da693cc15db07aac176f08958\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e1cf88de8c3c32f387615a27dc00b56472d3236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e1cf88de8c3c32f387615a27dc00b56472d3236"}]},"branch":"refs/heads/master"},"f1001deede958c49a60eb0a0f0e6960a057a0a2f":{"kind":"REWORK","_number":5,"created":"2019-05-24 00:19:39.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/27/654627/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/654627/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/654627/5"}}},"commit":{"parents":[{"commit":"104e6f051802dd2a66882473f783bf7cadede118","subject":"Merge \"Update release and branch to Stein for tripleo-upgrade.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/104e6f051802dd2a66882473f783bf7cadede118"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:42:08.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-24 00:18:03.000000000","tz":600},"subject":"Move openstack-infra to new namespaces","message":"Move openstack-infra to new namespaces\n\nChange-Id: I12d8738da634a76da693cc15db07aac176f08958\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1001deede958c49a60eb0a0f0e6960a057a0a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1001deede958c49a60eb0a0f0e6960a057a0a2f"}]},"branch":"refs/heads/master"},"da5bc141bcacff868e1402d74e782b947974728c":{"kind":"REWORK","_number":6,"created":"2019-05-28 14:53:58.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/27/654627/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/654627/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/654627/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/654627/6"}}},"commit":{"parents":[{"commit":"104e6f051802dd2a66882473f783bf7cadede118","subject":"Merge \"Update release and branch to Stein for tripleo-upgrade.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/104e6f051802dd2a66882473f783bf7cadede118"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-22 20:42:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-28 14:52:01.000000000","tz":-300},"subject":"Move openstack-infra to new namespaces","message":"Move openstack-infra to new namespaces\n\nChange-Id: I12d8738da634a76da693cc15db07aac176f08958\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da5bc141bcacff868e1402d74e782b947974728c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da5bc141bcacff868e1402d74e782b947974728c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
