)]}'
{"id":"openstack%2Freleases~655447","triplet_id":"openstack%2Freleases~master~I50de0cd52f2d43a2d8eb75fb82b06611b6ceb875","project":"openstack/releases","branch":"master","topic":"horizon-stein-15.1.0","hashtags":[],"change_id":"I50de0cd52f2d43a2d8eb75fb82b06611b6ceb875","subject":"Release Horizon Stein 15.1.0","status":"MERGED","created":"2019-04-24 13:54:11.000000000","updated":"2019-05-28 15:11:18.000000000","submitted":"2019-05-28 15:11:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655447-1559056278605-c481b760","meta_rev_id":"7f681c9113aff9583142061615319fee5e79fde6","_number":655447,"virtual_id_number":655447,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":2,"date":"2019-05-28 15:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-28 14:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-28 03:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":2,"date":"2019-05-28 01:21:32.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":1,"date":"2019-05-27 20:14:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-05-28 02:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-05-28 01:36:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-28 14:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"},{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 08:52:56.000000000","updated_by":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"reviewer":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"state":"REVIEWER"},{"updated":"2019-05-09 06:01:34.000000000","updated_by":{"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"},"reviewer":{"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"},"state":"REVIEWER"},{"updated":"2019-05-28 01:21:32.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-28 01:36:47.000000000","updated_by":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"reviewer":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-05-28 02:55:43.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-05-28 03:41:43.000000000","updated_by":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"reviewer":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"state":"REVIEWER"},{"updated":"2019-05-28 14:58:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-28 15:11:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2aac286d67e0d06fc4284d11f49076bc734fdecc","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-24 13:54:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05cfd83274c68c27139287bc0d7bf2dc52bbea82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:09:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/655447/1/check/openstack-tox-docs/77d5716/html/ : SUCCESS in 7m 38s\n- openstack-tox-validate http://logs.openstack.org/47/655447/1/check/openstack-tox-validate/9dddf36/ : FAILURE in 4m 17s\n- releases-tox-list-changes http://logs.openstack.org/47/655447/1/check/releases-tox-list-changes/b283d85/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0dcd262fe6cbb650bea0a95271c30fe6eb5886b0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-24 14:30:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1c3dda6117ae44d4b9f5b96afe67a640fe4863bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:42:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/655447/2/check/openstack-tox-docs/84e6cda/html/ : SUCCESS in 7m 35s\n- openstack-tox-validate http://logs.openstack.org/47/655447/2/check/openstack-tox-validate/89ad198/ : FAILURE in 5m 44s\n- releases-tox-list-changes http://logs.openstack.org/47/655447/2/check/releases-tox-list-changes/c6c56f6/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2b3551d188f74354b86bcb98712bb349634275aa","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-24 14:59:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b7010eb2b044fd033dfe1352599d77c6f1e0e143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:10:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/47/655447/3/check/openstack-tox-docs/208b74b/html/ : SUCCESS in 7m 02s\n- openstack-tox-validate http://logs.openstack.org/47/655447/3/check/openstack-tox-validate/34fbbb1/ : SUCCESS in 5m 29s\n- releases-tox-list-changes http://logs.openstack.org/47/655447/3/check/releases-tox-list-changes/bfb2d5c/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ee2a545fb334f8ad63dfe1dd0d3a5fcdff9ca710","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-04-24 15:59:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7eac1e7c891958cb8b95720c6aa1073738707081","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-06 14:25:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cdc36e719af0b6d37d15411e269b3ee24ce48b8e","author":{"_account_id":20147,"name":"Wei Li","email":"liwei@awcloud.com","username":"weili"},"date":"2019-05-08 08:52:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fa81fd74b1a321179dfcee9f7a472025e55b6880","author":{"_account_id":30398,"name":"xiehao","email":"xiehao@awcloud.com","username":"xiehao"},"date":"2019-05-09 06:01:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ec848ddbe29281fdef05f178453941e6944a2ce6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 22:37:32.000000000","message":"Patch Set 3: Code-Review-1\n\nThis release includes I0fc4d5db39080985b471fe2465b7078e37417e4d which bumps the minimum for python-cinderclient to \u003e\u003d 4.0.0.\n\nThis is a stable policy violation so at the very least this needs to be 15.1.0 and have a release note.\n\nAlso 4.0.0[1] is blocked on stein so the minimum needs to be \u003e\u003d4.0.1[2]\n\nPlease fix master, and backport to stein\n\n[1] https://opendev.org/openstack/requirements/src/branch/stable/stein/global-requirements.txt#L256\n[2] This may indicate a bug in requirements-check, or perhaps just a race","accounts_in_message":[],"_revision_number":3},{"id":"7f7d8b8412b0efafc6e6d7121cf24d567222be5a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-10 01:02:46.000000000","message":"Patch Set 3: -Code-Review\n\n\u003e This release includes I0fc4d5db39080985b471fe2465b7078e37417e4d which bumps the minimum for python-cinderclient to \u003e\u003d 4.0.0.\n\u003e \n\u003e This is a stable policy violation so at the very least this needs to be 15.1.0 and have a release note.\n\u003e \n\u003e Also 4.0.0[1] is blocked on stein so the minimum needs to be \u003e\u003d4.0.1[2]\n\u003e \n\u003e Please fix master, and backport to stein\n\u003e \n\u003e [1] https://opendev.org/openstack/requirements/src/branch/stable/stein/global-requirements.txt#L256\n\u003e [2] This may indicate a bug in requirements-check, or perhaps just a race\n\nAfter relooking the original bug 1824017, this looks like a combination bug of cinderclient and cinder microversion.\nCinder API 3.55 introduces a new validation and horizon uses Cinder API 3.58 (if available).\nThe corresponding horizon feature can work Cinder API \u003e\u003d 3.27. There is no need to use Cinder API \u003e\u003d 3.55. [3]\n\nThere seems several solutions:\n- bump the lower bound of cinderclient to 4.0.x and bump the horizon minor version to 15.1.0.\n- change horizon not to use Cinder API \u003e\u003d3.55 when creating a volume. 3.48 should work. This does not change cinderclient requirements.\n\nWhich sounds a better solution? Or other solutions?\n\nThis looks like a red flag to me that API consumer projects like horizon cannot handle the lower bound appropriately because every consumers cannot track all changes and know the minimumber version needs to be bumped. This kind of lower bound changes can easily happen in the current way of maintaining requirements.txt.\n\n[3] https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/api/microversions.py#L42","accounts_in_message":[],"_revision_number":3},{"id":"d987f85f0a0d8cd373a24690d705a18631fd8e00","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 15:54:01.000000000","message":"Patch Set 3: -Code-Review\n\nSince according to the bug some distros may have already packaged up this change, I would vote for making this 15.1.0 and keeping the raised l-c.","accounts_in_message":[],"_revision_number":3},{"id":"e226996564d58f2793df45d9fc5cb3c9fab0c1b3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-11 01:45:12.000000000","message":"Patch Set 3: Code-Review-1\n\nThanks for the input. I am preparing on the update as suggested in horizon side.","accounts_in_message":[],"_revision_number":3},{"id":"0ebf6013400084dcddab298a60d01ce777bcd268","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-27 09:55:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9e18f267167b5fd531e26242f6202dfb84245542","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-27 09:55:39.000000000","message":"Topic changed from horizon-stein-15.0.1 to horizon-stein-15.1.0","accounts_in_message":[],"_revision_number":4},{"id":"26f95da6628b618e0241a64d9821e7cb34ca400e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 10:07:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/47/655447/4/check/openstack-tox-docs/44f26ab/html/ : SUCCESS in 7m 10s\n- openstack-tox-validate http://logs.openstack.org/47/655447/4/check/openstack-tox-validate/3cc52bb/ : SUCCESS in 4m 31s\n- releases-tox-list-changes http://logs.openstack.org/47/655447/4/check/releases-tox-list-changes/ea1f015/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":4},{"id":"e11359e47337a259d746898ac64b82855296923f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-27 20:14:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f74cacdc82f33a31d46bb138d8bc7578726344df","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-28 01:21:32.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks.\n\nLooks good to me","accounts_in_message":[],"_revision_number":4},{"id":"ca7bc933492981b79ca7529bea532301846feecd","author":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"date":"2019-05-28 01:36:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"751faccff0c2fb3d8021e76bec6a7d257b51f945","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-28 02:55:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"04c08b7366b34cb9e42b648ec6653e464c236faa","author":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"date":"2019-05-28 03:41:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"eca4f44fb4435e590f0c161b6bd1fc3713de9043","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-28 14:58:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b1bc918fe159147bdcbab1f1c3f2e9f7953ac745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:58:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c414eb5ef533086ce97e7adeb7e86f4c2a126007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:11:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/47/655447/4/gate/openstack-tox-docs/0da7ec7/html/ : SUCCESS in 7m 05s\n- openstack-tox-validate http://logs.openstack.org/47/655447/4/gate/openstack-tox-validate/22d768a/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"7f681c9113aff9583142061615319fee5e79fde6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:11:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"874d82392a25ffae92af74ba3867201984768f81","revisions":{"0d673de1c9cffaef9c75d8c58261b0a391ac014d":{"kind":"REWORK","_number":1,"created":"2019-04-24 13:54:11.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/47/655447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/655447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/655447/1"}}},"commit":{"parents":[{"commit":"762e357c0d821c35b0373ac344d697807be031bf","subject":"Merge \"[Telemetry] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/762e357c0d821c35b0373ac344d697807be031bf"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 13:53:20.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 13:53:20.000000000","tz":180},"subject":"Release Horizon Stein 15.0.1","message":"Release Horizon Stein 15.0.1\n\nChange-Id: I50de0cd52f2d43a2d8eb75fb82b06611b6ceb875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d673de1c9cffaef9c75d8c58261b0a391ac014d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d673de1c9cffaef9c75d8c58261b0a391ac014d"}]},"branch":"refs/heads/master"},"206f42bf7eda11431148e7b983ae16c400a542b3":{"kind":"REWORK","_number":2,"created":"2019-04-24 14:30:39.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/47/655447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/655447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/655447/2"}}},"commit":{"parents":[{"commit":"762e357c0d821c35b0373ac344d697807be031bf","subject":"Merge \"[Telemetry] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/762e357c0d821c35b0373ac344d697807be031bf"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 13:53:20.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 14:30:35.000000000","tz":0},"subject":"Release Horizon Stein 15.0.1","message":"Release Horizon Stein 15.0.1\n\nChange-Id: I50de0cd52f2d43a2d8eb75fb82b06611b6ceb875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/206f42bf7eda11431148e7b983ae16c400a542b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/206f42bf7eda11431148e7b983ae16c400a542b3"}]},"branch":"refs/heads/master"},"8314dc680aaae384859e862aa053d19267be633c":{"kind":"REWORK","_number":3,"created":"2019-04-24 14:59:45.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/47/655447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/655447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/655447/3"}}},"commit":{"parents":[{"commit":"762e357c0d821c35b0373ac344d697807be031bf","subject":"Merge \"[Telemetry] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/762e357c0d821c35b0373ac344d697807be031bf"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 13:53:20.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 14:59:30.000000000","tz":180},"subject":"Release Horizon Stein 15.0.1","message":"Release Horizon Stein 15.0.1\n\nChange-Id: I50de0cd52f2d43a2d8eb75fb82b06611b6ceb875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8314dc680aaae384859e862aa053d19267be633c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8314dc680aaae384859e862aa053d19267be633c"}]},"branch":"refs/heads/master"},"874d82392a25ffae92af74ba3867201984768f81":{"kind":"REWORK","_number":4,"created":"2019-05-27 09:55:20.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/47/655447/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/655447/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/655447/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/655447/4"}}},"commit":{"parents":[{"commit":"762e357c0d821c35b0373ac344d697807be031bf","subject":"Merge \"[Telemetry] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/762e357c0d821c35b0373ac344d697807be031bf"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 13:53:20.000000000","tz":180},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-05-27 09:55:18.000000000","tz":540},"subject":"Release Horizon Stein 15.1.0","message":"Release Horizon Stein 15.1.0\n\nChange-Id: I50de0cd52f2d43a2d8eb75fb82b06611b6ceb875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/874d82392a25ffae92af74ba3867201984768f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/874d82392a25ffae92af74ba3867201984768f81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
