)]}'
{"id":"openstack%2Freleases~657072","triplet_id":"openstack%2Freleases~master~Ibefd2b96cb3409267a8ed6b5e68b5b52255577b6","project":"openstack/releases","branch":"master","topic":"idle-time-em","hashtags":[],"change_id":"Ibefd2b96cb3409267a8ed6b5e68b5b52255577b6","subject":"Modify Queens Extended Maintenance date","status":"MERGED","created":"2019-05-03 22:20:01.000000000","updated":"2019-06-24 14:15:42.000000000","submitted":"2019-06-24 14:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657072-1561385742293-18733654","meta_rev_id":"bb9fd4e300eb0599104ee127630464eed75ec8e5","_number":657072,"virtual_id_number":657072,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-06-20 12:59:09.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-06-24 14:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-20 13:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-06-24 14:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-24 14:01:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 21:40:14.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-11 01:14:00.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-06-20 13:08:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-24 14:01:17.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-24 14:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"537b08c9df2d11dbe3bb5eff3eec9adee8ce9de8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-05-03 22:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"874c6c904115921b31550df71b480d0e412c043b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 22:29:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/657072/1/check/openstack-tox-docs/d23e333/html/ : SUCCESS in 7m 08s\n- openstack-tox-validate http://logs.openstack.org/72/657072/1/check/openstack-tox-validate/708e1d3/ : SUCCESS in 4m 08s\n- releases-tox-list-changes http://logs.openstack.org/72/657072/1/check/releases-tox-list-changes/a8b2af5/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"af9984543016a131a4d030ce3d3374ef07ecb352","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 21:40:14.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks Elod!\n\nLooks good to me","accounts_in_message":[],"_revision_number":1},{"id":"2afa671aaf7a31eb5161af75aa99c8c83f83e322","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-11 01:14:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"177d2b72247aed18e5fd0b36ed85a442205f5148","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-31 13:45:05.000000000","message":"Patch Set 1: Code-Review+2\n\nWorks for me. Will wait to approve to give others a chance to review.","accounts_in_message":[],"_revision_number":1},{"id":"af96d7982c1a9909249c43b8de3d8d580ef511f9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-20 12:58:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96f7388552003b42868dab3781ec53660e6b5d84","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-20 12:59:09.000000000","message":"Patch Set 2:\n\nRebased.","accounts_in_message":[],"_revision_number":2},{"id":"54aa4b53e7dd8ee51a67e039aa1e8808dc9f877e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-20 13:08:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e62947cbaecfcacdf363c6bb9cff415fe6693b2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:18:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/72/657072/2/check/openstack-tox-py35/de9f958/ : SUCCESS in 14m 35s\n- openstack-tox-py36 http://logs.openstack.org/72/657072/2/check/openstack-tox-py36/d577708/ : SUCCESS in 6m 22s\n- openstack-tox-docs http://logs.openstack.org/72/657072/2/check/openstack-tox-docs/3f7e671/html/ : SUCCESS in 8m 24s\n- openstack-tox-validate http://logs.openstack.org/72/657072/2/check/openstack-tox-validate/d88604c/ : SUCCESS in 4m 17s\n- releases-tox-list-changes http://logs.openstack.org/72/657072/2/check/releases-tox-list-changes/a794dfd/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/72/657072/2/check/openstack-tox-pep8/4087d55/ : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ba00292d2112656c506c233e0bf395e94753d2d9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-24 14:01:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Elod!","accounts_in_message":[],"_revision_number":2},{"id":"57c4e7c2f30c392f09b0e22f125ed96c62ed0d22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:01:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a631739262221e76540e6864aa2b040f8669c482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:15:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"42d48bb2a6133d224355311906935c6439f26fa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:15:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/72/657072/2/gate/openstack-tox-py35/58dd9f2/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/72/657072/2/gate/openstack-tox-py36/efb2b71/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/72/657072/2/gate/openstack-tox-docs/52289d9/html/ : SUCCESS in 7m 57s\n- openstack-tox-validate http://logs.openstack.org/72/657072/2/gate/openstack-tox-validate/7e043ba/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be797b61677b1eafc0162bd664f09db54287ac81","revisions":{"1aff1d12f2c6fbee5dd577e59b02669a1d0fd5d6":{"kind":"REWORK","_number":1,"created":"2019-05-03 22:20:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/72/657072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/657072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/657072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/657072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/657072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/657072/1"}}},"commit":{"parents":[{"commit":"4936cae262e348c33f8844319421233acc2158a0","subject":"Merge \"[ironic] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4936cae262e348c33f8844319421233acc2158a0"}]}],"author":{"name":"Előd Illés","email":"elod.illes@ericsson.com","date":"2019-05-03 21:28:29.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@ericsson.com","date":"2019-05-03 21:28:29.000000000","tz":120},"subject":"Modify Queens Extended Maintenance date","message":"Modify Queens Extended Maintenance date\n\nAs it was discussed during Release Team meeting @ PTG the EM release\nshould happen somewhere around post release time.\n\nChange-Id: Ibefd2b96cb3409267a8ed6b5e68b5b52255577b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1aff1d12f2c6fbee5dd577e59b02669a1d0fd5d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1aff1d12f2c6fbee5dd577e59b02669a1d0fd5d6"}]},"branch":"refs/heads/master"},"be797b61677b1eafc0162bd664f09db54287ac81":{"kind":"REWORK","_number":2,"created":"2019-06-20 12:58:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/72/657072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/657072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/657072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/657072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/657072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/657072/2"}}},"commit":{"parents":[{"commit":"a9e488f400bdabadefacb0835d54947b936c2577","subject":"Merge \"Release puppet-manila for Stein, Rocky and Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9e488f400bdabadefacb0835d54947b936c2577"}]}],"author":{"name":"Előd Illés","email":"elod.illes@ericsson.com","date":"2019-05-03 21:28:29.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-06-20 12:56:50.000000000","tz":120},"subject":"Modify Queens Extended Maintenance date","message":"Modify Queens Extended Maintenance date\n\nAs it was discussed during Release Team meeting @ PTG the EM release\nshould happen somewhere around post release time.\n\nChange-Id: Ibefd2b96cb3409267a8ed6b5e68b5b52255577b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be797b61677b1eafc0162bd664f09db54287ac81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be797b61677b1eafc0162bd664f09db54287ac81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
