)]}'
{"id":"openstack%2Freleases~657897","triplet_id":"openstack%2Freleases~master~Iee7a609ef00e12dcee2a3f899ef531ffa68b8c72","project":"openstack/releases","branch":"master","topic":"rally_1_5_0","hashtags":[],"change_id":"Iee7a609ef00e12dcee2a3f899ef531ffa68b8c72","subject":"Propose a new release of rally","status":"MERGED","created":"2019-05-08 22:09:25.000000000","updated":"2019-05-09 22:12:15.000000000","submitted":"2019-05-09 22:12:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657897-1557439935386-5c776a74","meta_rev_id":"fab28eb8dd325b3e4a2ba1d8cf1649619f40abe4","_number":657897,"virtual_id_number":657897,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-05-09 22:12:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-05-09 22:02:26.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-09 15:23:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-09 06:19:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-05-09 22:02:26.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 06:19:46.000000000","updated_by":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"reviewer":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"state":"REVIEWER"},{"updated":"2019-05-09 15:23:32.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-09 22:02:26.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-09 22:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42cf08e7a7a9fe9f9fa3b1d2c85d75acfb857efd","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-08 22:09:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0b40b4603b90aaf571e174884d1d5647f69cb8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 22:18:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/657897/1/check/openstack-tox-docs/cfbd8c6/html/ : SUCCESS in 7m 43s\n- openstack-tox-validate http://logs.openstack.org/97/657897/1/check/openstack-tox-validate/d9a13e1/ : SUCCESS in 4m 45s\n- releases-tox-list-changes http://logs.openstack.org/97/657897/1/check/releases-tox-list-changes/097bb02/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f6447431e499fd02ba41e71c1d08a68211ad6be7","author":{"_account_id":23294,"name":"LiuFu","email":"liufu@awcloud.com","username":"LiuFu"},"date":"2019-05-09 06:19:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"035580eb85595558e4bcfe768fd3599ab20b56c5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-09 15:23:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9d8ce038eb0d920ed6beec28a6add2cc64b7741","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 22:02:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9dce5c73f99b53c16722528a111c973e5e6d8f5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 22:02:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d04a92b261ced0686e0e44c5ae5a8bb239181cf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 22:12:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65c8a1b7d27e1ca8ba8727d2fa6c9760912cceb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 22:12:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/657897/1/gate/openstack-tox-docs/058b87f/html/ : SUCCESS in 7m 52s\n- openstack-tox-validate http://logs.openstack.org/97/657897/1/gate/openstack-tox-validate/8cdfa51/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59699a676ab5b5d94867fe794fce17c9a10f2535","revisions":{"59699a676ab5b5d94867fe794fce17c9a10f2535":{"kind":"REWORK","_number":1,"created":"2019-05-08 22:09:25.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/97/657897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/657897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/657897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/657897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/657897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/657897/1"}}},"commit":{"parents":[{"commit":"03ccc759a8a99874e983e4b8b1ef5161d22ebd3c","subject":"Merge \"oslo.db 4.40.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/03ccc759a8a99874e983e4b8b1ef5161d22ebd3c"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-08 22:09:21.000000000","tz":-420},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-08 22:09:21.000000000","tz":-420},"subject":"Propose a new release of rally","message":"Propose a new release of rally\n\nChange-Id: Iee7a609ef00e12dcee2a3f899ef531ffa68b8c72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59699a676ab5b5d94867fe794fce17c9a10f2535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59699a676ab5b5d94867fe794fce17c9a10f2535"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
