)]}'
{"id":"openstack%2Freleases~658561","triplet_id":"openstack%2Freleases~master~I0d46b1a4e00c1775ea5e38d39e87bed99f6bbb2e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I0d46b1a4e00c1775ea5e38d39e87bed99f6bbb2e","subject":"Rename review.openstack.org to review.opendev.org","status":"MERGED","created":"2019-05-11 20:33:27.000000000","updated":"2019-05-16 06:27:20.000000000","submitted":"2019-05-16 06:27:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658561-1557988040887-a4b81bb3","meta_rev_id":"54846aabc02413a26c809b5229f0676e05be52f6","_number":658561,"virtual_id_number":658561,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-16 06:27:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-05-16 06:14:58.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-13 11:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-05-15 09:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-05-16 06:14:58.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 11:51:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-15 09:39:14.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-16 06:14:58.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-16 06:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cca09ee3ed752ad2bb464c77e36f0a96b7364af","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-11 20:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab5f400441b325d6afaf96fae6db36885947e050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 20:50:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/61/658561/1/check/openstack-tox-py35/7783240/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/61/658561/1/check/openstack-tox-py36/d26f480/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/61/658561/1/check/openstack-tox-docs/86d6197/html/ : SUCCESS in 6m 58s\n- openstack-tox-validate http://logs.openstack.org/61/658561/1/check/openstack-tox-validate/4233a9b/ : SUCCESS in 4m 38s\n- releases-tox-list-changes http://logs.openstack.org/61/658561/1/check/releases-tox-list-changes/7b733d7/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/61/658561/1/check/openstack-tox-pep8/2c7e770/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"287eab2da7e37becaa9fae92db0716ede68f8f3b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-13 11:51:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ba5b9e1770120eae828cf8bfd1009ffbe2162e3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-15 09:39:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ba6a9512c3a573fc34f67f5b5cff33bff4a8ffc","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 06:14:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b71015b04dcb1dfc9d677f5c0e13d27a7276e86b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:15:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b497d38891973707fb4d87f824cd4fbe0490c30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:27:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54846aabc02413a26c809b5229f0676e05be52f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:27:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/61/658561/1/gate/openstack-tox-py35/6a97473/ : SUCCESS in 6m 10s\n- openstack-tox-py36 http://logs.openstack.org/61/658561/1/gate/openstack-tox-py36/35f2eab/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/61/658561/1/gate/openstack-tox-docs/1e03a52/html/ : SUCCESS in 7m 05s\n- openstack-tox-validate http://logs.openstack.org/61/658561/1/gate/openstack-tox-validate/00654e7/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7000617626ad770c30c19bdc601ce293c521550","revisions":{"e7000617626ad770c30c19bdc601ce293c521550":{"kind":"REWORK","_number":1,"created":"2019-05-11 20:33:27.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/61/658561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/658561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/658561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/658561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/658561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/658561/1"}}},"commit":{"parents":[{"commit":"425d613cb6095579aa813b28e50f01bc7c835fe9","subject":"Merge \"[sahara] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/425d613cb6095579aa813b28e50f01bc7c835fe9"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-11 20:31:25.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-11 20:33:25.000000000","tz":480},"subject":"Rename review.openstack.org to review.opendev.org","message":"Rename review.openstack.org to review.opendev.org\n\nThere are many references to review.openstack.org, and while the\nredirect should work, we can also go ahead and fix them.\n\nChange-Id: I0d46b1a4e00c1775ea5e38d39e87bed99f6bbb2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7000617626ad770c30c19bdc601ce293c521550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7000617626ad770c30c19bdc601ce293c521550"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
