)]}'
{"id":"openstack%2Freleases~659291","triplet_id":"openstack%2Freleases~master~Ib8923b23be71d872458952e37567c39abbc6b267","project":"openstack/releases","branch":"master","topic":"novaclient-14.0.0","hashtags":[],"change_id":"Ib8923b23be71d872458952e37567c39abbc6b267","subject":"train: release python-novaclient 14.0.0","status":"MERGED","created":"2019-05-15 13:16:39.000000000","updated":"2019-05-16 06:42:45.000000000","submitted":"2019-05-16 06:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659291-1557988965402-ec8a5968","meta_rev_id":"df049624e0269c142c7fb1928dbcf9dbe4aed451","_number":659291,"virtual_id_number":659291,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-05-16 06:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-15 18:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-16 06:30:02.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-05-15 20:00:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-16 06:30:02.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 17:21:41.000000000","updated_by":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"reviewer":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"state":"REVIEWER"},{"updated":"2019-05-15 18:59:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-15 20:00:23.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-16 06:30:02.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-16 06:42:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"241615e11be18942043477ccb83c89d1ea7772a1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-15 13:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"635571088a3be314aefb4e572ebe3dd0d6217b64","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-15 13:16:57.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting for https://review.opendev.org/#/c/659202/ to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f179b638e78ffaff156fd86e01a816983d0de4f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:27:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/659291/1/check/openstack-tox-docs/48ae5fe/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/91/659291/1/check/openstack-tox-validate/485074f/ : FAILURE in 4m 17s\n- releases-tox-list-changes http://logs.openstack.org/91/659291/1/check/releases-tox-list-changes/04f1938/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"aaaceb42ca2011d682df3a4f2db993df17eb4599","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 15:11:08.000000000","message":"Patch Set 1:\n\nlooks like the failure is legit because of missing a new line : http://logs.openstack.org/91/659291/1/check/openstack-tox-validate/485074f/job-output.txt.gz#_2019-05-15_13_23_24_492606","accounts_in_message":[],"_revision_number":1},{"id":"7eb0cd1fa8cfc1994f767f6e7488b5eaaf5d97e0","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 15:14:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"11dcd6f0370dd5f61e62d67260a8ffedb76f8c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:24:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/659291/2/check/openstack-tox-docs/4656e68/html/ : SUCCESS in 8m 53s\n- openstack-tox-validate http://logs.openstack.org/91/659291/2/check/openstack-tox-validate/52c9048/ : SUCCESS in 4m 07s\n- releases-tox-list-changes http://logs.openstack.org/91/659291/2/check/releases-tox-list-changes/a31c5f9/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"9288f765f1db2e72138fa5d5b5327e4c275e363a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-15 16:36:05.000000000","message":"Patch Set 2: Workflow-1\n\nWaiting for the dependency to merge since I\u0027d like to get that in the release.","accounts_in_message":[],"_revision_number":2},{"id":"b70d45ce7743f0d6d8efdb2866b640b8a3a1bc04","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 17:20:09.000000000","message":"Patch Set 2:\n\n\u003e Waiting for the dependency to merge since I\u0027d like to get that in\n \u003e the release.\n\nits merged now.","accounts_in_message":[],"_revision_number":2},{"id":"23e11df98345c112df56f68c06b45a6d4b45c564","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 17:21:41.000000000","message":"Patch Set 2: Code-Review+1\n\nnot that I have ever reviewed a release patch, but it looks similar to the older release patches (and because I need it for my osc change) so +1!","accounts_in_message":[],"_revision_number":2},{"id":"24377b53aa6ee2cac2b9f8b890713b75a6f1ca1e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-15 18:11:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23a7450d176f2fd4297513e469812df5cb3b8099","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 18:23:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/659291/3/check/openstack-tox-docs/7613c9b/html/ : SUCCESS in 6m 42s\n- openstack-tox-validate http://logs.openstack.org/91/659291/3/check/openstack-tox-validate/543a552/ : SUCCESS in 3m 53s\n- releases-tox-list-changes http://logs.openstack.org/91/659291/3/check/releases-tox-list-changes/756b8fc/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"cea4bbca0e15f7a02f4b589438ae8a8f05e4ae05","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-15 18:59:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"eb13a4e67feb06b8131341e9a83c18b8588c6989","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-15 20:00:23.000000000","message":"Patch Set 3: Code-Review+1\n\nHash checks out, no other major open items that are close to merging, agree with major version bump.\n\n/me wields PTL ack gavel.","accounts_in_message":[],"_revision_number":3},{"id":"e24f77b6c70f7e04c2f4a442c13fd0487d82d74d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 06:30:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc6782068ca9c3616e9abbad2e7745b25021a678","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:30:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6e0131905ae4b574c50e25688e28c1241c34ac15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:42:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"76c04a657a12a26ca826f774f3061ad3abc46a89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:42:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/659291/3/gate/openstack-tox-docs/b11ac12/html/ : SUCCESS in 6m 49s\n- openstack-tox-validate http://logs.openstack.org/91/659291/3/gate/openstack-tox-validate/4e553f1/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9dc4bb13415122da5d1df8c960c7f1feab6676d4","revisions":{"ce9cb4f17b927789bbb95a9e2894f4d0b69d8d93":{"kind":"REWORK","_number":1,"created":"2019-05-15 13:16:39.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/91/659291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/659291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/659291/1"}}},"commit":{"parents":[{"commit":"9b66dd3f39cd186187d7fedf7e53b612792d3b74","subject":"Merge \"Add Cycle Highlight Deadline to Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b66dd3f39cd186187d7fedf7e53b612792d3b74"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-15 13:16:05.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-15 13:16:05.000000000","tz":-240},"subject":"train: release python-novaclient 14.0.0","message":"train: release python-novaclient 14.0.0\n\nThis is a major version bump because of change:\n\na7ea3fb0 Remove deprecated options\n\nDepends-On: https://review.opendev.org/659202/\nChange-Id: Ib8923b23be71d872458952e37567c39abbc6b267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce9cb4f17b927789bbb95a9e2894f4d0b69d8d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce9cb4f17b927789bbb95a9e2894f4d0b69d8d93"}]},"branch":"refs/heads/master"},"f3f4d6982866326ea156d1396d71185f29481d4c":{"kind":"REWORK","_number":2,"created":"2019-05-15 15:14:24.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/91/659291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/659291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/659291/2"}}},"commit":{"parents":[{"commit":"9b66dd3f39cd186187d7fedf7e53b612792d3b74","subject":"Merge \"Add Cycle Highlight Deadline to Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b66dd3f39cd186187d7fedf7e53b612792d3b74"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-15 13:16:05.000000000","tz":-240},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-05-15 15:14:14.000000000","tz":120},"subject":"train: release python-novaclient 14.0.0","message":"train: release python-novaclient 14.0.0\n\nThis is a major version bump because of change:\n\na7ea3fb0 Remove deprecated options\n\nDepends-On: https://review.opendev.org/659202/\nChange-Id: Ib8923b23be71d872458952e37567c39abbc6b267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3f4d6982866326ea156d1396d71185f29481d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3f4d6982866326ea156d1396d71185f29481d4c"}]},"branch":"refs/heads/master"},"9dc4bb13415122da5d1df8c960c7f1feab6676d4":{"kind":"REWORK","_number":3,"created":"2019-05-15 18:11:12.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/91/659291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/659291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/659291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/659291/3"}}},"commit":{"parents":[{"commit":"9b66dd3f39cd186187d7fedf7e53b612792d3b74","subject":"Merge \"Add Cycle Highlight Deadline to Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b66dd3f39cd186187d7fedf7e53b612792d3b74"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-15 13:16:05.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-15 18:10:57.000000000","tz":-240},"subject":"train: release python-novaclient 14.0.0","message":"train: release python-novaclient 14.0.0\n\nThis is a major version bump because of change:\n\na7ea3fb0 Remove deprecated options\n\nChange-Id: Ib8923b23be71d872458952e37567c39abbc6b267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9dc4bb13415122da5d1df8c960c7f1feab6676d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9dc4bb13415122da5d1df8c960c7f1feab6676d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
