)]}'
{"id":"openstack%2Freleases~659330","triplet_id":"openstack%2Freleases~master~I54f0542c4652597bbf6b7d8b1b4f9b37511ed96e","project":"openstack/releases","branch":"master","topic":"cinder-train-schedule","hashtags":[],"change_id":"I54f0542c4652597bbf6b7d8b1b4f9b37511ed96e","subject":"Add Cinder Specific Events for Train","status":"MERGED","created":"2019-05-15 15:56:11.000000000","updated":"2019-05-16 13:30:10.000000000","submitted":"2019-05-16 13:30:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659330-1558013410062-a4b81bb3","meta_rev_id":"08642d0b82ee2b7f911f00310483660b38183f17","_number":659330,"virtual_id_number":659330,"owner":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-05-16 13:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-16 13:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-16 08:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-16 13:22:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 06:26:41.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-16 08:45:57.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2019-05-16 08:52:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-16 13:22:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-16 13:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17583b4ecf7054ac901f03468efff3994659c414","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-05-15 15:56:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0fbf6239f148a2a2ec932a13f88d9d769b5793e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 16:17:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/659330/1/check/openstack-tox-docs/b0fa9d3/html/ : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0aa3be56ec6cf7750ea5b45b586fdf1b6e4ef6ab","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-05-15 16:22:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"24cdb30612c76030f3acc4b7ad562f094aa5e490","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-05-15 18:53:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3cd4d7ba7860ff34534194ead8b3b8649a97338","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:01:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/659330/2/check/openstack-tox-docs/046b92e/html/ : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"21f64ab1853d456f33ce9728d3d01d735a8bb5ec","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 06:25:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0cb461b1cfdfff6caa6993c07905f967ccb8b398","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 06:26:41.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good to me.\n\nThanks Jay.  I hope you don\u0027t mind my inline fixes","accounts_in_message":[],"_revision_number":3},{"id":"e26ed0521c169232eb770dfafe422a38ca1d6557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:43:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/659330/3/check/openstack-tox-docs/3f8c566/html/ : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b66158c6b2f461a9257d9e4fbd55a2785b58d0eb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 08:15:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fdfa14297623076fecb5b1716e130b2d20706564","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:23:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/659330/4/check/openstack-tox-docs/a2da96e/html/ : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":4},{"id":"7cc0c8b4cd5fcd69fa54dba0c4950ef81d8098a4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-05-16 08:45:57.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks Tony, LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"389aa91f79fd09e4d81a45eadfa5e39f43c61a87","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-16 08:52:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"43a6234bd5ff4cc2f1017b40904b42bf195adf8f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 13:22:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3dd17a6df6833861ad231983e1960a99beaef690","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:22:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9a73b6d59d96c36738b7050854cbe627c0715a48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:30:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/659330/4/gate/openstack-tox-docs/4d4ac4a/html/ : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":4},{"id":"30fb263602ae34e85982166a41ebe3f2f21240c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:30:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6439fd081ec8bc7540b63213eee4fb2069d92485","revisions":{"67971ee92a205fcdc35c162b600e920dc1b7eaef":{"kind":"REWORK","_number":1,"created":"2019-05-15 15:56:11.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/30/659330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/30/659330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/30/659330/1"}}},"commit":{"parents":[{"commit":"9b66dd3f39cd186187d7fedf7e53b612792d3b74","subject":"Merge \"Add Cycle Highlight Deadline to Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b66dd3f39cd186187d7fedf7e53b612792d3b74"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-05-15 15:55:18.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-05-15 15:55:18.000000000","tz":-300},"subject":"Add Cinder Specific Events for Train","message":"Add Cinder Specific Events for Train\n\nThis patch adds the Cinder Specific events for\nthe Train release cycle.\n\nChange-Id: I54f0542c4652597bbf6b7d8b1b4f9b37511ed96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67971ee92a205fcdc35c162b600e920dc1b7eaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67971ee92a205fcdc35c162b600e920dc1b7eaef"}]},"branch":"refs/heads/master"},"c63f9e182ce8dc92fc221c79d95d794aebc87deb":{"kind":"REWORK","_number":2,"created":"2019-05-15 18:53:36.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/30/659330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/30/659330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/30/659330/2"}}},"commit":{"parents":[{"commit":"9b66dd3f39cd186187d7fedf7e53b612792d3b74","subject":"Merge \"Add Cycle Highlight Deadline to Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b66dd3f39cd186187d7fedf7e53b612792d3b74"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-05-15 15:55:18.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-05-15 18:53:26.000000000","tz":-300},"subject":"Add Cinder Specific Events for Train","message":"Add Cinder Specific Events for Train\n\nThis patch adds the Cinder Specific events for\nthe Train release cycle.\n\nChange-Id: I54f0542c4652597bbf6b7d8b1b4f9b37511ed96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c63f9e182ce8dc92fc221c79d95d794aebc87deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c63f9e182ce8dc92fc221c79d95d794aebc87deb"}]},"branch":"refs/heads/master"},"b74b3d1e7a3925b4083df7e8e68b47d790bf54a6":{"kind":"REWORK","_number":3,"created":"2019-05-16 06:25:12.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/30/659330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/30/659330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/30/659330/3"}}},"commit":{"parents":[{"commit":"9b66dd3f39cd186187d7fedf7e53b612792d3b74","subject":"Merge \"Add Cycle Highlight Deadline to Schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b66dd3f39cd186187d7fedf7e53b612792d3b74"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-05-15 15:55:18.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-16 06:25:07.000000000","tz":0},"subject":"Add Cinder Specific Events for Train","message":"Add Cinder Specific Events for Train\n\nThis patch adds the Cinder Specific events for\nthe Train release cycle.\n\nChange-Id: I54f0542c4652597bbf6b7d8b1b4f9b37511ed96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b74b3d1e7a3925b4083df7e8e68b47d790bf54a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b74b3d1e7a3925b4083df7e8e68b47d790bf54a6"}]},"branch":"refs/heads/master"},"6439fd081ec8bc7540b63213eee4fb2069d92485":{"kind":"REWORK","_number":4,"created":"2019-05-16 08:15:28.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/30/659330/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/30/659330/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/30/659330/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/30/659330/4"}}},"commit":{"parents":[{"commit":"85efd90025e2a9212d4c471f6bba0ede96331de2","subject":"Merge \"[Rally] Release a nwe version with a fix to previous release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/85efd90025e2a9212d4c471f6bba0ede96331de2"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2019-05-15 15:55:18.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-16 08:15:05.000000000","tz":600},"subject":"Add Cinder Specific Events for Train","message":"Add Cinder Specific Events for Train\n\nThis patch adds the Cinder Specific events for\nthe Train release cycle.\n\nChange-Id: I54f0542c4652597bbf6b7d8b1b4f9b37511ed96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6439fd081ec8bc7540b63213eee4fb2069d92485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6439fd081ec8bc7540b63213eee4fb2069d92485"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
