)]}'
{"id":"openstack%2Freleases~660046","triplet_id":"openstack%2Freleases~master~I4e64d0cf890f3b2f62da552c5b604426b7764d8e","project":"openstack/releases","branch":"master","topic":"ironic-lib","hashtags":[],"change_id":"I4e64d0cf890f3b2f62da552c5b604426b7764d8e","subject":"Release ironic-lib 2.17.0 for Train","status":"MERGED","created":"2019-05-20 08:29:15.000000000","updated":"2019-05-21 17:08:06.000000000","submitted":"2019-05-21 17:08:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660046-1558458486130-db3b04f4","meta_rev_id":"0dd07307fbf50431c5153c8dfe50539661ef7330","_number":660046,"virtual_id_number":660046,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-21 17:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-21 16:58:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-21 16:58:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 16:58:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-21 17:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b4e55e2ec2e546bebaa435b5dc185a3991a7db3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-20 08:29:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85d9a50be44d40c2058dcaaa1d7dfafae78f7ec6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 08:41:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/660046/1/check/openstack-tox-docs/cc52183/html/ : SUCCESS in 7m 01s\n- openstack-tox-validate http://logs.openstack.org/46/660046/1/check/openstack-tox-validate/ee7de94/ : SUCCESS in 3m 58s\n- releases-tox-list-changes http://logs.openstack.org/46/660046/1/check/releases-tox-list-changes/3efa1b2/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"836693227606766b629ea4be1d8fb41ebf4dbb78","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-21 16:58:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61cef9d1c9248a5d8a2beaf2a4c1db119684fc4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:58:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b95cd78c789e914a5e61ac88f4a6006c6d10bb58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:08:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/660046/1/gate/openstack-tox-docs/109b02e/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/46/660046/1/gate/openstack-tox-validate/625d32f/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"bf3520fb67dd5df01935a6b6c6f8095046dd3ea3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:08:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88e86de140011fd4b9c0faa92469371c0dc00324","revisions":{"88e86de140011fd4b9c0faa92469371c0dc00324":{"kind":"REWORK","_number":1,"created":"2019-05-20 08:29:15.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/660046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/660046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/660046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/660046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/660046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/660046/1"}}},"commit":{"parents":[{"commit":"07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc","subject":"Merge \"[tricircle] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-20 08:26:48.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-20 08:26:48.000000000","tz":120},"subject":"Release ironic-lib 2.17.0 for Train","message":"Release ironic-lib 2.17.0 for Train\n\nChange-Id: I4e64d0cf890f3b2f62da552c5b604426b7764d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88e86de140011fd4b9c0faa92469371c0dc00324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88e86de140011fd4b9c0faa92469371c0dc00324"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
