)]}'
{"id":"openstack%2Freleases~660444","triplet_id":"openstack%2Freleases~master~I948ea8514cf7b72cb06e64f956f2b93685b23028","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I948ea8514cf7b72cb06e64f956f2b93685b23028","subject":"Update release and branch to Stein for tripleo-upgrade.","status":"MERGED","created":"2019-05-21 16:09:41.000000000","updated":"2019-05-23 23:47:58.000000000","submitted":"2019-05-23 23:47:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660444-1558655278301-1fb69e20","meta_rev_id":"fe43ca9d9a2c03a9e66001dacd4032ee75fea4a1","_number":660444,"virtual_id_number":660444,"owner":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-05-22 13:37:38.000000000","_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-23 23:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"recommended":{"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-22 06:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},{"value":2,"date":"2019-05-23 05:59:17.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":1,"date":"2019-05-23 14:19:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-22 13:27:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},{"value":1,"date":"2019-05-23 23:40:13.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 16:09:59.000000000","updated_by":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"reviewer":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"state":"REVIEWER"},{"updated":"2019-05-21 16:10:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-21 16:10:25.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-21 16:10:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-22 06:57:47.000000000","updated_by":{"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},"reviewer":{"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},"state":"REVIEWER"},{"updated":"2019-05-22 13:27:49.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-23 14:19:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-23 23:40:13.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-23 23:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1cf35001d3f4750a27f57e9b841c29935d1bf7e2","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-05-21 16:09:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8aa6632b13ae427f77d2c90d0d5c5edd718442a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-21 16:13:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc99b711f2b69449edc25f76ca9e1b9cc5b68e57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:22:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/44/660444/1/check/openstack-tox-docs/4daea72/ : POST_FAILURE in 4m 23s\n- openstack-tox-validate http://logs.openstack.org/44/660444/1/check/openstack-tox-validate/427c298/ : FAILURE in 4m 17s\n- releases-tox-list-changes http://logs.openstack.org/44/660444/1/check/releases-tox-list-changes/72690a0/ : FAILURE in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"58f45c2e51705f6ff5e367576fe6cdeae45e0139","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-21 16:25:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ed9d9efdd60af1bf833869668f66f2a67b6ea50","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-05-22 06:40:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c53f306a36b242a301dbb4e04183d3248d146323","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 06:48:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/660444/2/check/openstack-tox-docs/2f17093/html/ : SUCCESS in 6m 42s\n- openstack-tox-validate http://logs.openstack.org/44/660444/2/check/openstack-tox-validate/fd41843/ : SUCCESS in 3m 34s\n- releases-tox-list-changes http://logs.openstack.org/44/660444/2/check/releases-tox-list-changes/f5e1bcf/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1610184a8323f16d1fcd337d144b7dac19d48b05","author":{"_account_id":21537,"name":"Yurii Prokulevych","email":"yprokule@redhat.com","username":"yprokule"},"date":"2019-05-22 06:57:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5bdd9958fe57933c410e4be5c9c43d87e701b545","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 13:27:49.000000000","message":"Patch Set 2: Code-Review+1\n\nI can\u0027t really judge on the sha to use, but it seems the rest is correct.","accounts_in_message":[],"_revision_number":2},{"id":"6cc9e5982d6276ae3795bee5da8ccb2783d4d6e9","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-05-22 13:37:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3f03e68b96b0ed3106272c0786bf0dc50f096a4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 05:59:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed7ee63285290177fdd587a132f577f00c236389","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 05:59:27.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"79b490022e1ad1f9418cb1ef7c22baf46c224dbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 05:59:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f01866f5dbcd5c849e86999e690ad9c4a212feb8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:00:31.000000000","message":"Patch Set 2:\n\nOoops\n\nI\u0027m pretty sure Alex is okay with this but we don\u0027t actually have his +1","accounts_in_message":[],"_revision_number":2},{"id":"7215badfc373af5742eca9a7ec6a7b6a0afd8322","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:25:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/660444/2/gate/openstack-tox-docs/c2851f4/html/ : SUCCESS in 7m 18s\n- openstack-tox-validate http://logs.openstack.org/44/660444/2/gate/openstack-tox-validate/e628a5d/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"719d0e7e7d011026e55edfb1bb0d788ea321b0e8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-23 14:19:26.000000000","message":"Patch Set 2: Code-Review+1\n\nyes would have been fine :D","accounts_in_message":[],"_revision_number":2},{"id":"3f9ea2a951c14ff34c5166fc0c78236b1f8ef06b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 23:40:13.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e612ab3af96279696e423f8c0e73b8ad6c407edb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 23:40:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aaab8a78652fd5b234082f90ebc276ab8b54c2b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 23:47:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/660444/2/gate/openstack-tox-docs/afe02d8/html/ : SUCCESS in 6m 48s\n- openstack-tox-validate http://logs.openstack.org/44/660444/2/gate/openstack-tox-validate/91eed24/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"fe43ca9d9a2c03a9e66001dacd4032ee75fea4a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 23:47:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a190ac10d74dfc7d8df2839583c2a9955f5fe252","revisions":{"5bc89f6b70a7a7ff723772511b78d999a9562de7":{"kind":"REWORK","_number":1,"created":"2019-05-21 16:09:41.000000000","uploader":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"ref":"refs/changes/44/660444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/660444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/660444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/660444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/660444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/660444/1"}}},"commit":{"parents":[{"commit":"07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc","subject":"Merge \"[tricircle] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc"}]}],"author":{"name":"Jose Luis Franco Arza","email":"jfrancoa@redhat.com","date":"2019-05-21 16:09:29.000000000","tz":120},"committer":{"name":"Jose Luis Franco Arza","email":"jfrancoa@redhat.com","date":"2019-05-21 16:09:29.000000000","tz":120},"subject":"Update release and branch to Stein for tripleo-upgrade.","message":"Update release and branch to Stein for tripleo-upgrade.\n\n- Tag: 4.0.0\n- Branch: stable/stein\n\nChange-Id: I948ea8514cf7b72cb06e64f956f2b93685b23028\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5bc89f6b70a7a7ff723772511b78d999a9562de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5bc89f6b70a7a7ff723772511b78d999a9562de7"}]},"branch":"refs/heads/master"},"a190ac10d74dfc7d8df2839583c2a9955f5fe252":{"kind":"REWORK","_number":2,"created":"2019-05-22 06:40:22.000000000","uploader":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"ref":"refs/changes/44/660444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/660444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/660444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/660444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/660444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/660444/2"}}},"commit":{"parents":[{"commit":"07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc","subject":"Merge \"[tricircle] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07d7ba4a1faa27e9cb2c310813ccd8dcca8d59bc"}]}],"author":{"name":"Jose Luis Franco Arza","email":"jfrancoa@redhat.com","date":"2019-05-21 16:09:29.000000000","tz":120},"committer":{"name":"Jose Luis Franco Arza","email":"jfrancoa@redhat.com","date":"2019-05-22 06:40:13.000000000","tz":120},"subject":"Update release and branch to Stein for tripleo-upgrade.","message":"Update release and branch to Stein for tripleo-upgrade.\n\n- Tag: 4.0.0\n- Branch: stable/stein\n\nChange-Id: I948ea8514cf7b72cb06e64f956f2b93685b23028\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a190ac10d74dfc7d8df2839583c2a9955f5fe252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a190ac10d74dfc7d8df2839583c2a9955f5fe252"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
