)]}'
{"id":"openstack%2Freleases~660553","triplet_id":"openstack%2Freleases~master~Ife030f8ee7b5d204b054f99e920a675f7d92da69","project":"openstack/releases","branch":"master","topic":"constraints-updates","hashtags":[],"change_id":"Ife030f8ee7b5d204b054f99e920a675f7d92da69","subject":"Switch constraints files over to the new opendev infrastructure","status":"MERGED","created":"2019-05-22 03:45:59.000000000","updated":"2019-05-28 15:09:19.000000000","submitted":"2019-05-28 15:09:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660553-1559056159021-c481b760","meta_rev_id":"f659fa0957cf548a747b2fa472a796ca4ef9a074","_number":660553,"virtual_id_number":660553,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-28 15:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-22 15:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-22 16:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2019-05-24 10:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-05-22 14:04:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-28 14:55:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 14:04:49.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-22 16:37:59.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-05-24 10:04:48.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-28 14:55:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-28 15:09:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b9132d4c3722634deba86014b5a5ee74b8848cd","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-22 03:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1458238996aaeff1e4d49e794458aab88db78106","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 04:12:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/53/660553/1/check/openstack-tox-py35/6b58cd8/ : FAILURE in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/53/660553/1/check/openstack-tox-py36/10c96f8/ : FAILURE in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/53/660553/1/check/openstack-tox-docs/71b9a84/html/ : SUCCESS in 6m 48s\n- openstack-tox-validate http://logs.openstack.org/53/660553/1/check/openstack-tox-validate/306aee9/ : SUCCESS in 3m 47s\n- releases-tox-list-changes http://logs.openstack.org/53/660553/1/check/releases-tox-list-changes/fecf366/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/53/660553/1/check/openstack-tox-pep8/37d92f1/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6b1dc494e0d4051d40af9bb7511095cdc53a064e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-22 04:34:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fea0fb078ce8b2eddeb0656ff5662b6e5c05703d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 04:43:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/53/660553/2/check/openstack-tox-py35/9bec9fc/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/53/660553/2/check/openstack-tox-py36/358aed8/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/53/660553/2/check/openstack-tox-docs/ca63c64/html/ : SUCCESS in 6m 15s\n- openstack-tox-validate http://logs.openstack.org/53/660553/2/check/openstack-tox-validate/e056019/ : SUCCESS in 3m 56s\n- releases-tox-list-changes http://logs.openstack.org/53/660553/2/check/releases-tox-list-changes/24d6afa/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/53/660553/2/check/openstack-tox-pep8/bdc2e57/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"fad2d5480d2c7edc506d7bbdfc93e5f189d9bb05","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 14:04:49.000000000","message":"Patch Set 2: Code-Review+1\n\nI suppose -em branches could indeed have floating requirements instead of using tags. This looks good to me, but I am not 100% sure I have the right insights.","accounts_in_message":[],"_revision_number":2},{"id":"0ae595ae4daba003c38767adefb427188a34fe74","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-22 15:47:37.000000000","message":"Patch Set 2: Code-Review+2\n\nAppears correct to me.","accounts_in_message":[],"_revision_number":2},{"id":"0cb1600e65f714ec467a1963f5b928def44f85d6","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-05-22 16:37:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"82f88db0418b51a4ff26f19789df8ab58d49a5e1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-24 10:04:48.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM but a Friday sounds like a bad time to approve this.","accounts_in_message":[],"_revision_number":2},{"id":"fa4f656a4703a099e0aaaf02825e6bae7aee2460","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-28 14:55:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"020f5202e9fbd05b5e4e1b5a357e3d2ad5d638c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:55:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b13dd46f96e276f1666ee5f97231c2ff273d6a3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:09:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/53/660553/2/gate/openstack-tox-py35/af09255/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/53/660553/2/gate/openstack-tox-py36/ab2ffb0/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/53/660553/2/gate/openstack-tox-docs/dcf1b4c/html/ : SUCCESS in 7m 05s\n- openstack-tox-validate http://logs.openstack.org/53/660553/2/gate/openstack-tox-validate/1e9b2d9/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8c10344b3be33d3479e4f55c6f9ed4489e5f0e70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:09:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5801fbdc7781e5a09aa75f3b2ea4cbfd211bf5c9","revisions":{"db796c69d1dcd7f1153fe9e277978d6aee511f9e":{"kind":"REWORK","_number":1,"created":"2019-05-22 03:45:59.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/53/660553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/660553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/660553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/660553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/660553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/660553/1"}}},"commit":{"parents":[{"commit":"b50b346d63e4ecdaed3d94b66b7a4605546782cc","subject":"Merge \"Oslo releases for 2019-05-21\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b50b346d63e4ecdaed3d94b66b7a4605546782cc"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-22 03:39:46.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-22 03:39:46.000000000","tz":600},"subject":"Switch constraints files over to the new opendev infrastructure","message":"Switch constraints files over to the new opendev infrastructure\n\nCurrently files that point to EOL releases don\u0027t work as expected due to\ngitea needed to differentiate between tags and branches.  Rather than\nfix, and rely on multiple redirects lets just craft 301\u0027s that poin to\nthe correct gitea url.\n\nThis means we now need to know when a target is a branch or tag but\nthat\u0027s pretty simple to intuit given our deliverable structure.\n\nChange-Id: Ife030f8ee7b5d204b054f99e920a675f7d92da69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db796c69d1dcd7f1153fe9e277978d6aee511f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db796c69d1dcd7f1153fe9e277978d6aee511f9e"}]},"branch":"refs/heads/master"},"5801fbdc7781e5a09aa75f3b2ea4cbfd211bf5c9":{"kind":"REWORK","_number":2,"created":"2019-05-22 04:34:40.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/53/660553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/660553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/660553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/660553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/660553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/660553/2"}}},"commit":{"parents":[{"commit":"b50b346d63e4ecdaed3d94b66b7a4605546782cc","subject":"Merge \"Oslo releases for 2019-05-21\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b50b346d63e4ecdaed3d94b66b7a4605546782cc"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-22 03:39:46.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-22 04:33:37.000000000","tz":600},"subject":"Switch constraints files over to the new opendev infrastructure","message":"Switch constraints files over to the new opendev infrastructure\n\nCurrently constraint redirectiones that point to EOL releases don\u0027t work\nas expected due to gitea needed to differentiate between tags and\nbranches.  Rather than fix, and rely on multiple redirects lets just\ncraft 301\u0027s that point to the correct gitea url.\n\nThis means we now need to know when a target is a branch or tag but\nthat\u0027s pretty simple to intuit given our deliverable structure.\n\nChange-Id: Ife030f8ee7b5d204b054f99e920a675f7d92da69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5801fbdc7781e5a09aa75f3b2ea4cbfd211bf5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5801fbdc7781e5a09aa75f3b2ea4cbfd211bf5c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
