)]}'
{"id":"openstack%2Freleases~660590","triplet_id":"openstack%2Freleases~master~I35bfc96e7cfeb3a3f1f663aed5a132702f2b4465","project":"openstack/releases","branch":"master","topic":"inspector-rocky","hashtags":[],"change_id":"I35bfc96e7cfeb3a3f1f663aed5a132702f2b4465","subject":"Release ironic-inspector 8.0.3 for Rocky (security issue)","status":"MERGED","created":"2019-05-22 07:45:23.000000000","updated":"2019-05-22 15:56:55.000000000","submitted":"2019-05-22 15:56:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660590-1558540615643-bd4b8c07","meta_rev_id":"0805062f3f3157aa1cd107b65a86f7210f3db1dd","_number":660590,"virtual_id_number":660590,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-22 15:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-22 15:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-22 13:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-22 15:28:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 13:24:10.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-22 15:28:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-22 15:56:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4da78e43819cad471ec7237daf45623cc1174d7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-22 07:45:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0878d5ab45a35083122507bde025aa0b92842bc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 08:05:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/660590/1/check/openstack-tox-docs/40fed31/html/ : SUCCESS in 6m 14s\n- openstack-tox-validate http://logs.openstack.org/90/660590/1/check/openstack-tox-validate/a48fe25/ : SUCCESS in 4m 54s\n- releases-tox-list-changes http://logs.openstack.org/90/660590/1/check/releases-tox-list-changes/563d6d6/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6b6f4111321daf5ee26f5e8cb900c770d791402b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 13:24:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97f89f5cbe14e1a7da6dc43553030e7e67e1f6cf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-22 15:28:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nApproving ahead of normal stable review day since there are just a couple minure fixes in addition to the security bug fix.","accounts_in_message":[],"_revision_number":1},{"id":"ba9eed8cb62623c06d614fdbf8d36c5e4e98279b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 15:28:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7972974db8ac92588307662374f5bc11589866d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 15:56:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0046f3a67df48f125fc3cc69f911c0fd651db299","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 15:56:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/660590/1/gate/openstack-tox-docs/f69fabc/html/ : SUCCESS in 8m 56s\n- openstack-tox-validate http://logs.openstack.org/90/660590/1/gate/openstack-tox-validate/7da6247/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e55a1056d02d9f6dd6ecdd187fa0c57e1a1cd635","revisions":{"e55a1056d02d9f6dd6ecdd187fa0c57e1a1cd635":{"kind":"REWORK","_number":1,"created":"2019-05-22 07:45:23.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/90/660590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/660590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/660590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/660590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/660590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/660590/1"}}},"commit":{"parents":[{"commit":"b50b346d63e4ecdaed3d94b66b7a4605546782cc","subject":"Merge \"Oslo releases for 2019-05-21\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b50b346d63e4ecdaed3d94b66b7a4605546782cc"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-22 07:45:13.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2019-05-22 07:45:13.000000000","tz":120},"subject":"Release ironic-inspector 8.0.3 for Rocky (security issue)","message":"Release ironic-inspector 8.0.3 for Rocky (security issue)\n\nChange-Id: I35bfc96e7cfeb3a3f1f663aed5a132702f2b4465\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e55a1056d02d9f6dd6ecdd187fa0c57e1a1cd635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e55a1056d02d9f6dd6ecdd187fa0c57e1a1cd635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
