)]}'
{"id":"openstack%2Freleases~660712","triplet_id":"openstack%2Freleases~master~I47283bda07454e12bb8814c9ce43e722731b8e84","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I47283bda07454e12bb8814c9ce43e722731b8e84","subject":"Release networking-hyperv 7.2.0 for Stein","status":"MERGED","created":"2019-05-22 13:24:40.000000000","updated":"2019-05-24 00:26:08.000000000","submitted":"2019-05-24 00:26:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660712-1558657568334-e37cffb5","meta_rev_id":"369c38fef4ee880d4b92b3a352b5709045ded627","_number":660712,"virtual_id_number":660712,"owner":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-24 00:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-05-24 00:14:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-05-24 00:14:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 00:14:53.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-24 00:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7b903ade02ad4b2d58dedfabe5c83a30714b789","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2019-05-22 13:24:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"718e971e8ee15240b911390db30fe1ae1c8cef70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 13:39:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/660712/1/check/openstack-tox-docs/bec5b1b/html/ : SUCCESS in 7m 14s\n- openstack-tox-validate http://logs.openstack.org/12/660712/1/check/openstack-tox-validate/4a2069b/ : SUCCESS in 4m 15s\n- releases-tox-list-changes http://logs.openstack.org/12/660712/1/check/releases-tox-list-changes/f56b1de/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c2012b09a594e3c4c6ba39f455d86dd1fa3155de","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:02:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a97d7453c3a47ac4d4ec3cf037a3b173b715b52","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2019-05-23 06:53:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f0deb47c91235210a5e342608ca2f16ec9803e45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 07:09:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/660712/2/check/openstack-tox-docs/57b7225/html/ : SUCCESS in 6m 43s\n- openstack-tox-validate http://logs.openstack.org/12/660712/2/check/openstack-tox-validate/83dd81e/ : SUCCESS in 3m 55s\n- releases-tox-list-changes http://logs.openstack.org/12/660712/2/check/releases-tox-list-changes/d71f8ab/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"01e8d94ebaa56df29f02c41222e5d8692da30b46","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 23:39:06.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8dba2dc1a48d8f5cf2895b9bea8771968462246c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 00:03:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/660712/3/check/openstack-tox-docs/f795bf0/html/ : SUCCESS in 7m 12s\n- openstack-tox-validate http://logs.openstack.org/12/660712/3/check/openstack-tox-validate/d60fc86/ : SUCCESS in 5m 30s\n- releases-tox-list-changes http://logs.openstack.org/12/660712/3/check/releases-tox-list-changes/a7207df/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"c0d47f7410ebe57e6c43f1e1b974cdade5e311b1","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 00:14:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks.\n\nLooks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"7254375e9bcd6d6e74c84197420691ad0eb41e03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 00:15:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f31cdad3104d6ebf9a1a0e18eaf62454c19d8958","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 00:26:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"369c38fef4ee880d4b92b3a352b5709045ded627","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 00:26:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/660712/3/gate/openstack-tox-docs/dd8324b/html/ : SUCCESS in 7m 23s\n- openstack-tox-validate http://logs.openstack.org/12/660712/3/gate/openstack-tox-validate/d1c56b5/ : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"41372315ea0391c9f48502d056317be3527297dd","revisions":{"a89820520013d9a7e22cf1c6f254a2f50289f165":{"kind":"REWORK","_number":1,"created":"2019-05-22 13:24:40.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/12/660712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/660712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/660712/1"}}},"commit":{"parents":[{"commit":"b50b346d63e4ecdaed3d94b66b7a4605546782cc","subject":"Merge \"Oslo releases for 2019-05-21\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b50b346d63e4ecdaed3d94b66b7a4605546782cc"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2019-05-21 16:20:47.000000000","tz":-420},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2019-05-21 16:20:47.000000000","tz":-420},"subject":"Release networking-hyperv 7.1.1 for Stein","message":"Release networking-hyperv 7.1.1 for Stein\n\nChange-Id: I47283bda07454e12bb8814c9ce43e722731b8e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a89820520013d9a7e22cf1c6f254a2f50289f165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a89820520013d9a7e22cf1c6f254a2f50289f165"}]},"branch":"refs/heads/master"},"6a1bb926d44ed31c028b22fa7e385284d31bab49":{"kind":"REWORK","_number":2,"created":"2019-05-23 06:53:57.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/12/660712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/660712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/660712/2"}}},"commit":{"parents":[{"commit":"b50b346d63e4ecdaed3d94b66b7a4605546782cc","subject":"Merge \"Oslo releases for 2019-05-21\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b50b346d63e4ecdaed3d94b66b7a4605546782cc"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2019-05-21 16:20:47.000000000","tz":-420},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2019-05-23 06:53:53.000000000","tz":0},"subject":"Release networking-hyperv 7.1.1 for Stein","message":"Release networking-hyperv 7.1.1 for Stein\n\nChange-Id: I47283bda07454e12bb8814c9ce43e722731b8e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a1bb926d44ed31c028b22fa7e385284d31bab49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a1bb926d44ed31c028b22fa7e385284d31bab49"}]},"branch":"refs/heads/master"},"41372315ea0391c9f48502d056317be3527297dd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-23 23:39:06.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/12/660712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/660712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/660712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/660712/3"}}},"commit":{"parents":[{"commit":"b50b346d63e4ecdaed3d94b66b7a4605546782cc","subject":"Merge \"Oslo releases for 2019-05-21\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b50b346d63e4ecdaed3d94b66b7a4605546782cc"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2019-05-21 16:20:47.000000000","tz":-420},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-23 23:39:03.000000000","tz":0},"subject":"Release networking-hyperv 7.2.0 for Stein","message":"Release networking-hyperv 7.2.0 for Stein\n\nChange-Id: I47283bda07454e12bb8814c9ce43e722731b8e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/41372315ea0391c9f48502d056317be3527297dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/41372315ea0391c9f48502d056317be3527297dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
