)]}'
{"id":"openstack%2Freleases~661376","triplet_id":"openstack%2Freleases~master~I2105bda1eefe7a58a626081aab50894ff79955ac","project":"openstack/releases","branch":"master","topic":"nova-stein-19.0.1","hashtags":[],"change_id":"I2105bda1eefe7a58a626081aab50894ff79955ac","subject":"nova: release stein 19.0.1","status":"MERGED","created":"2019-05-24 18:41:31.000000000","updated":"2019-06-06 06:15:31.000000000","submitted":"2019-06-06 06:15:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661376-1559801731181-c7054fac","meta_rev_id":"7c6d6361409158c0c4de916720243b7ebdec749d","_number":661376,"virtual_id_number":661376,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-06-06 06:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-06 06:00:33.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-04 19:26:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-05 16:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-06-05 10:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-06 06:00:33.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 13:50:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-04 19:26:06.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-05 10:47:52.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-06-05 16:49:16.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-06 06:00:33.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-06 06:15:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56c30a463fdf4fa8813bbab7b1c50f5cdddffaeb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-24 18:41:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f71c0d325dd687725018053a37e34cc1a23174f5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-24 18:44:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"66571079be2bd7135a63ca7632c506d209ec4ce3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 18:59:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/661376/2/check/openstack-tox-docs/8ba479c/html/ : SUCCESS in 7m 13s\n- openstack-tox-validate http://logs.openstack.org/76/661376/2/check/openstack-tox-validate/eff58ac/ : SUCCESS in 7m 53s\n- releases-tox-list-changes http://logs.openstack.org/76/661376/2/check/releases-tox-list-changes/fa55eaa/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"abc27f72edc3e0ae43347c74716ff3e1ce3c73b4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-24 19:03:34.000000000","message":"Patch Set 2: Workflow-1\n\nWe may want a known issue release note for bug 1829062 and running nova-api with wsgi + eventlet monkeypatching.","accounts_in_message":[],"_revision_number":2},{"id":"6db9882793890e82ae7e2a24bc90e4844c96c396","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-24 20:06:13.000000000","message":"Patch Set 2:\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-May/006617.html","accounts_in_message":[],"_revision_number":2},{"id":"de7afee5ca513d7c304bd6dcb8fe8446ad93b5aa","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-29 20:24:17.000000000","message":"Patch Set 2: -Workflow\n\nIt sounds like a reno is a good idea but we don\u0027t know enough yet what it should really say or what the suggested actions would be, so rather than delay this release, which already has a lot of unreleased changes, let\u0027s just get this in and do a reno later if needed.","accounts_in_message":[],"_revision_number":2},{"id":"438981d2360044a63d99bfd2ec62e4a414d04f4a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-30 01:22:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ff03978a2cfe8e094c2001e501e7a898b37d128d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-30 14:32:22.000000000","message":"Patch Set 2: Code-Review+1\n\nCould go to dc7eb62c27 at this point, but this is fine.","accounts_in_message":[],"_revision_number":2},{"id":"942a352a34ad989244b6924cdcb06527971c51e1","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-31 11:10:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49208997530a775f316687e601b00d00869068d3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-31 13:50:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c28d2af8b0cb659447f8b3d5493a27d4286f96df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:50:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c0697e9fde0a9da54c29b243aa633960d226a567","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-31 13:54:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee0d75c7c1e2fbbdab72681f193772c8e599cd23","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-31 13:58:57.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f41f758a37f072aa5f8bbacd0a902c7fd7b42e2c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-31 13:59:19.000000000","message":"Patch Set 3: Workflow-1\n\nSorry, realized that the rocky version of https://review.opendev.org/#/c/656304/ has merged and so we need to release stein with https://review.opendev.org/#/c/656304/ before we do a rocky release.","accounts_in_message":[],"_revision_number":3},{"id":"0bb82209d7f238abc80ac4c3ba02b4d1910ab0e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 14:13:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/661376/3/check/openstack-tox-docs/9d7419f/html/ : SUCCESS in 7m 37s\n- openstack-tox-validate http://logs.openstack.org/76/661376/3/check/openstack-tox-validate/1f367d0/ : SUCCESS in 7m 53s\n- releases-tox-list-changes http://logs.openstack.org/76/661376/3/check/releases-tox-list-changes/b505f80/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":3},{"id":"d8dd4bbeb5849044475b359a391cdb2b083402d0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 17:57:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1dd55c04c3ea241a52c06c5176ac59ad51fd5eef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 18:10:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/661376/4/check/openstack-tox-docs/07c0c4c/html/ : SUCCESS in 7m 24s\n- openstack-tox-validate http://logs.openstack.org/76/661376/4/check/openstack-tox-validate/72acf8b/ : SUCCESS in 7m 07s\n- releases-tox-list-changes http://logs.openstack.org/76/661376/4/check/releases-tox-list-changes/31af7ab/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"38b408a797248fd8e3e6fb6e05f96237096af324","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-04 19:26:06.000000000","message":"Patch Set 4: Code-Review+1\n\nHash points to current tip of stein branch, which incorporates the commit mriedem mentions in his comment at PS3.","accounts_in_message":[],"_revision_number":4},{"id":"27ca4747dd5416f731491fab4bddb5adfb2b797c","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-06-05 10:47:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f11c6736a942b9787a6e9b61b5e954540c709928","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-05 16:46:27.000000000","message":"Patch Set 4:\n\n\u003e Hash points to current tip of stein branch, which incorporates the\n \u003e commit mriedem mentions in his comment at PS3.\n\nThanks for checking!","accounts_in_message":[],"_revision_number":4},{"id":"6dc15f63ce5d7eed428dd385bd9d8451d161847c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-05 16:49:16.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM.\nSide note: I am impressed everytime I see a list of bugfixes this big. Team is doing great job ! :)","accounts_in_message":[],"_revision_number":4},{"id":"97f2b38f378a21998a2a66f0efc80cd9043e2d3d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-06 06:00:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f0acd13fba887b05b26b8375782ccfd0c45ed4b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:00:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6965a98077874344a855fef9684205311d242725","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:15:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/661376/4/gate/openstack-tox-docs/87d1829/html/ : SUCCESS in 7m 01s\n- openstack-tox-validate http://logs.openstack.org/76/661376/4/gate/openstack-tox-validate/ed8a2d0/ : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":4},{"id":"fd0857082279ad8a0756dfee9030d5f1eb0d04ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:15:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d0779d66418a839522b98369e2fe0bd545e6dbce","revisions":{"a5ed83b592faba0bdfab6796966434aa64cb6140":{"kind":"REWORK","_number":1,"created":"2019-05-24 18:41:31.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/76/661376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/661376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/661376/1"}}},"commit":{"parents":[{"commit":"1d756fe83c634408555ae68398e124fdf620b071","subject":"Merge \"Release networking-hyperv 7.2.0 for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d756fe83c634408555ae68398e124fdf620b071"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-24 18:39:34.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-24 18:39:34.000000000","tz":-240},"subject":"nova: release stein 19.0.1","message":"nova: release stein 19.0.1\n\nThis release contains a few fixes for regressions found\nafter the Stein 19.0.0 GA plus lots of other bug fixes.\n\n$ git log --oneline --no-merges 19.0.0..origin/stable/stein\nc094a6ef0c (bug/1827628) Disable limit if affinity(anti)/same(different)host is requested\n4ff46ff767 Restore connection_info after live migration rollback\nf1280ab849 libvirt: Do not reraise DiskNotFound exceptions during resize\ne13513010f Fix target used in nova.policy.check_is_admin\n2f7b1031a9 Stop logging traceback when skipping quiesce\n8330ab4e61 Skip _exclude_nested_providers() if not nested\nd250834734 Fix assert methods in unit tests\n61fef49b15 Fix retry of instance_update_and_get_original\nd7d7f11543 (bug/1821594) Add functional confirm_migration_error test\ncfc3fa4908 Only call _fill_provider_mapping if claim succeeds\n1c774315a5 Handle placement error during re-schedule\nc38a821902 Fix exception type in test_boot_reschedule_fill_provider_mapping_raises\n07a1a8ff7d Reproduce bug #1819460 in functional test\na2b1ef5f11 Log when port resource is leaked during port delete\n91282f879c Use migration_status during volume migrating and retyping\naf07ddb734 (bug/1825034-stein) Exclude fake marker instance when listing servers\n840f109afc Add regression test for bug 1825034\n1d4f64b190 (review/lee_yarwood/bug/1821798) Include all network devices in nova diagnostics\n048d5b790f libvirt: Always disconnect volumes after libvirtError exceptions\nb3dedefcc5 libvirt: Stop ignoring unknown libvirtError exceptions during volume attach\n3285210add (review/matt_riedemann/bug/1825020) Fix volume-backed resize with a smaller disk flavor\nb123b05a48 Fix help for ironic.peer_list config\na1db4507ac Add functional regression recreate test for bug 1825020\nc6432ac021 libvirt: Avoid using os-brick encryptors when device_path isn\u0027t provided\nc280d747fb AZ list performance optimization: avoid double service list DB fetch\n7cd816f43a Do not perform port update in case of baremetal instance.\n222f462444 Pass on region when we don\u0027t have a valid ironic endpoint\n870e5bcfb6 Create request spec, build request and mappings in one transaction\ne6ca3838d9 Fix {min|max}_version in ironic Adapter setup\nb3b1e6d4da OpenDev Migration Patch\n7500c1910c libvirt: set device address tag only if setting disk unit\n2bbd7f4a0c Change a log level for overwriting allocation\nd773e3b146 libvirt: disconnect volume when encryption fails\nfe5c8a517a Do not log a warning about not using compute monitors\nef25fff5ec doc: Capitalize keystone domain name\n5f515060f6 (review/matt_riedemann/bug/1821594) Delete allocations even if _confirm_resize raises\n17e6ca7272 Don\u0027t warn on network-vif-unplugged event during live migration\n972d4e0eb3 Error out migration when confirm_resize fails\n6a5be4ae25 Fix bug preventing forbidden traits from working\nc97466456f Adding tests to demonstrate bug #1821824\nb1b28169af Move create of ComputeAPI object in websocketproxy\n76dfb9d0b6 (bug/1669054-stein) Do not persist RequestSpec.ignore_hosts\n20c1414945 Add functional regression test for bug 1669054\nbcb42a43a9 Temporarily mutate migration object in finish_revert_resize\n99b3f44eb1 Add retry_on_deadlock to migration_update DB API\n7ccea385b1 doc: Fix openstack CLI command\n26e5991283 (bug/1819963-stein) Update instance.availability_zone on revertResize\n0b2345442d Add functional recreate test for bug 1819963\neec39762cf (review/elod_illes/bug/1819430) Fix incomplete instance data returned after build failure\na98fb711b5 Use Selection object to fill request group mapping\nae40ceba84 Override the \u0027get\u0027 method in DriverBlockDevice class\nd0f540742e Eventlet monkey patching should be as early as possible\n\nChange-Id: I2105bda1eefe7a58a626081aab50894ff79955ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5ed83b592faba0bdfab6796966434aa64cb6140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5ed83b592faba0bdfab6796966434aa64cb6140"}]},"branch":"refs/heads/master"},"b78ef1d350a3adbbf6d94f9b0ad60ef71b958b13":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-24 18:44:58.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/76/661376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/661376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/661376/2"}}},"commit":{"parents":[{"commit":"1d756fe83c634408555ae68398e124fdf620b071","subject":"Merge \"Release networking-hyperv 7.2.0 for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d756fe83c634408555ae68398e124fdf620b071"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-24 18:39:34.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-24 18:43:14.000000000","tz":-240},"subject":"nova: release stein 19.0.1","message":"nova: release stein 19.0.1\n\nThis release contains a few fixes for regressions found\nafter the Stein 19.0.0 GA plus lots of other bug fixes.\n\n$ git log --oneline --no-merges --no-decorate 19.0.0..origin/stable/stein\nc094a6ef0c Disable limit if affinity(anti)/same(different)host is requested\n4ff46ff767 Restore connection_info after live migration rollback\nf1280ab849 libvirt: Do not reraise DiskNotFound exceptions during resize\ne13513010f Fix target used in nova.policy.check_is_admin\n2f7b1031a9 Stop logging traceback when skipping quiesce\n8330ab4e61 Skip _exclude_nested_providers() if not nested\nd250834734 Fix assert methods in unit tests\n61fef49b15 Fix retry of instance_update_and_get_original\nd7d7f11543 Add functional confirm_migration_error test\ncfc3fa4908 Only call _fill_provider_mapping if claim succeeds\n1c774315a5 Handle placement error during re-schedule\nc38a821902 Fix exception type in test_boot_reschedule_fill_provider_mapping_raises\n07a1a8ff7d Reproduce bug #1819460 in functional test\na2b1ef5f11 Log when port resource is leaked during port delete\n91282f879c Use migration_status during volume migrating and retyping\naf07ddb734 Exclude fake marker instance when listing servers\n840f109afc Add regression test for bug 1825034\n1d4f64b190 Include all network devices in nova diagnostics\n048d5b790f libvirt: Always disconnect volumes after libvirtError exceptions\nb3dedefcc5 libvirt: Stop ignoring unknown libvirtError exceptions during volume attach\n3285210add Fix volume-backed resize with a smaller disk flavor\nb123b05a48 Fix help for ironic.peer_list config\na1db4507ac Add functional regression recreate test for bug 1825020\nc6432ac021 libvirt: Avoid using os-brick encryptors when device_path isn\u0027t provided\nc280d747fb AZ list performance optimization: avoid double service list DB fetch\n7cd816f43a Do not perform port update in case of baremetal instance.\n222f462444 Pass on region when we don\u0027t have a valid ironic endpoint\n870e5bcfb6 Create request spec, build request and mappings in one transaction\ne6ca3838d9 Fix {min|max}_version in ironic Adapter setup\nb3b1e6d4da OpenDev Migration Patch\n7500c1910c libvirt: set device address tag only if setting disk unit\n2bbd7f4a0c Change a log level for overwriting allocation\nd773e3b146 libvirt: disconnect volume when encryption fails\nfe5c8a517a Do not log a warning about not using compute monitors\nef25fff5ec doc: Capitalize keystone domain name\n5f515060f6 Delete allocations even if _confirm_resize raises\n17e6ca7272 Don\u0027t warn on network-vif-unplugged event during live migration\n972d4e0eb3 Error out migration when confirm_resize fails\n6a5be4ae25 Fix bug preventing forbidden traits from working\nc97466456f Adding tests to demonstrate bug #1821824\nb1b28169af Move create of ComputeAPI object in websocketproxy\n76dfb9d0b6 Do not persist RequestSpec.ignore_hosts\n20c1414945 Add functional regression test for bug 1669054\nbcb42a43a9 Temporarily mutate migration object in finish_revert_resize\n99b3f44eb1 Add retry_on_deadlock to migration_update DB API\n7ccea385b1 doc: Fix openstack CLI command\n26e5991283 Update instance.availability_zone on revertResize\n0b2345442d Add functional recreate test for bug 1819963\neec39762cf Fix incomplete instance data returned after build failure\na98fb711b5 Use Selection object to fill request group mapping\nae40ceba84 Override the \u0027get\u0027 method in DriverBlockDevice class\nd0f540742e Eventlet monkey patching should be as early as possible\n\nChange-Id: I2105bda1eefe7a58a626081aab50894ff79955ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b78ef1d350a3adbbf6d94f9b0ad60ef71b958b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b78ef1d350a3adbbf6d94f9b0ad60ef71b958b13"}]},"branch":"refs/heads/master"},"fbdb30d21e731ff393c8882ec8c2d9b16ae9ad95":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-31 13:58:57.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/76/661376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/661376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/661376/3"}}},"commit":{"parents":[{"commit":"1d756fe83c634408555ae68398e124fdf620b071","subject":"Merge \"Release networking-hyperv 7.2.0 for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d756fe83c634408555ae68398e124fdf620b071"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-24 18:39:34.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-31 13:58:55.000000000","tz":0},"subject":"nova: release stein 19.0.1","message":"nova: release stein 19.0.1\n\nDepends-On: https://review.opendev.org/#/c/656304/\n\nThis release contains a few fixes for regressions found\nafter the Stein 19.0.0 GA plus lots of other bug fixes.\n\n$ git log --oneline --no-merges --no-decorate 19.0.0..origin/stable/stein\nc094a6ef0c Disable limit if affinity(anti)/same(different)host is requested\n4ff46ff767 Restore connection_info after live migration rollback\nf1280ab849 libvirt: Do not reraise DiskNotFound exceptions during resize\ne13513010f Fix target used in nova.policy.check_is_admin\n2f7b1031a9 Stop logging traceback when skipping quiesce\n8330ab4e61 Skip _exclude_nested_providers() if not nested\nd250834734 Fix assert methods in unit tests\n61fef49b15 Fix retry of instance_update_and_get_original\nd7d7f11543 Add functional confirm_migration_error test\ncfc3fa4908 Only call _fill_provider_mapping if claim succeeds\n1c774315a5 Handle placement error during re-schedule\nc38a821902 Fix exception type in test_boot_reschedule_fill_provider_mapping_raises\n07a1a8ff7d Reproduce bug #1819460 in functional test\na2b1ef5f11 Log when port resource is leaked during port delete\n91282f879c Use migration_status during volume migrating and retyping\naf07ddb734 Exclude fake marker instance when listing servers\n840f109afc Add regression test for bug 1825034\n1d4f64b190 Include all network devices in nova diagnostics\n048d5b790f libvirt: Always disconnect volumes after libvirtError exceptions\nb3dedefcc5 libvirt: Stop ignoring unknown libvirtError exceptions during volume attach\n3285210add Fix volume-backed resize with a smaller disk flavor\nb123b05a48 Fix help for ironic.peer_list config\na1db4507ac Add functional regression recreate test for bug 1825020\nc6432ac021 libvirt: Avoid using os-brick encryptors when device_path isn\u0027t provided\nc280d747fb AZ list performance optimization: avoid double service list DB fetch\n7cd816f43a Do not perform port update in case of baremetal instance.\n222f462444 Pass on region when we don\u0027t have a valid ironic endpoint\n870e5bcfb6 Create request spec, build request and mappings in one transaction\ne6ca3838d9 Fix {min|max}_version in ironic Adapter setup\nb3b1e6d4da OpenDev Migration Patch\n7500c1910c libvirt: set device address tag only if setting disk unit\n2bbd7f4a0c Change a log level for overwriting allocation\nd773e3b146 libvirt: disconnect volume when encryption fails\nfe5c8a517a Do not log a warning about not using compute monitors\nef25fff5ec doc: Capitalize keystone domain name\n5f515060f6 Delete allocations even if _confirm_resize raises\n17e6ca7272 Don\u0027t warn on network-vif-unplugged event during live migration\n972d4e0eb3 Error out migration when confirm_resize fails\n6a5be4ae25 Fix bug preventing forbidden traits from working\nc97466456f Adding tests to demonstrate bug #1821824\nb1b28169af Move create of ComputeAPI object in websocketproxy\n76dfb9d0b6 Do not persist RequestSpec.ignore_hosts\n20c1414945 Add functional regression test for bug 1669054\nbcb42a43a9 Temporarily mutate migration object in finish_revert_resize\n99b3f44eb1 Add retry_on_deadlock to migration_update DB API\n7ccea385b1 doc: Fix openstack CLI command\n26e5991283 Update instance.availability_zone on revertResize\n0b2345442d Add functional recreate test for bug 1819963\neec39762cf Fix incomplete instance data returned after build failure\na98fb711b5 Use Selection object to fill request group mapping\nae40ceba84 Override the \u0027get\u0027 method in DriverBlockDevice class\nd0f540742e Eventlet monkey patching should be as early as possible\n\nChange-Id: I2105bda1eefe7a58a626081aab50894ff79955ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fbdb30d21e731ff393c8882ec8c2d9b16ae9ad95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fbdb30d21e731ff393c8882ec8c2d9b16ae9ad95"}]},"branch":"refs/heads/master"},"d0779d66418a839522b98369e2fe0bd545e6dbce":{"kind":"REWORK","_number":4,"created":"2019-06-04 17:57:37.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/76/661376/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/661376/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/661376/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/661376/4"}}},"commit":{"parents":[{"commit":"1d756fe83c634408555ae68398e124fdf620b071","subject":"Merge \"Release networking-hyperv 7.2.0 for Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d756fe83c634408555ae68398e124fdf620b071"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-24 18:39:34.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-04 17:57:26.000000000","tz":-240},"subject":"nova: release stein 19.0.1","message":"nova: release stein 19.0.1\n\nThis release contains a few fixes for regressions found\nafter the Stein 19.0.0 GA plus lots of other bug fixes.\n\nChange-Id: I2105bda1eefe7a58a626081aab50894ff79955ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0779d66418a839522b98369e2fe0bd545e6dbce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0779d66418a839522b98369e2fe0bd545e6dbce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
