)]}'
{"id":"openstack%2Freleases~661926","triplet_id":"openstack%2Freleases~master~Ie6743fa52434abf8ba0f0aa4db741552e6ccfb95","project":"openstack/releases","branch":"master","topic":"bug/1828265","hashtags":[],"change_id":"Ie6743fa52434abf8ba0f0aa4db741552e6ccfb95","subject":"Release openstackdocstheme 1.30.0","status":"MERGED","created":"2019-05-29 09:48:26.000000000","updated":"2019-05-30 01:34:03.000000000","submitted":"2019-05-30 01:34:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661926-1559180043246-3d446f26","meta_rev_id":"6190332f7353de5dea0d065e52ee8281ffe899b6","_number":661926,"virtual_id_number":661926,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-05-29 09:48:46.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-05-30 01:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-05-30 01:21:07.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-05-29 09:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-05-29 13:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-05-29 22:12:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-29 10:36:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-05-30 01:21:07.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 09:57:02.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-29 10:36:50.000000000","updated_by":{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"},"reviewer":{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"},"state":"REVIEWER"},{"updated":"2019-05-29 13:59:19.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-29 22:12:27.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-30 01:21:07.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-30 01:34:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20e7676f384d324f0a4ee8fbffd547297ad09506","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-29 09:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5808a1912ff778917673c88b4d2fae6b6ad95e87","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-29 09:48:46.000000000","message":"Patch Set 1:\n\nThis is needed to get [1] in the wild\n\n[1] https://review.opendev.org/#/c/661344/","accounts_in_message":[],"_revision_number":1},{"id":"1608bd2ddb0e6362f4578087bf46076371a9351c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-29 09:57:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"725fa8c42014d7fb0eb17cd1fd5ca30f5fb12ca3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 10:09:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/26/661926/1/check/openstack-tox-docs/4f8796a/html/ : SUCCESS in 6m 49s\n- openstack-tox-validate http://logs.openstack.org/26/661926/1/check/openstack-tox-validate/d7f828c/ : SUCCESS in 4m 52s\n- releases-tox-list-changes http://logs.openstack.org/26/661926/1/check/releases-tox-list-changes/3ea55ed/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4843f3faf414010007db807576cec976dc739d81","author":{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"},"date":"2019-05-29 10:36:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9b62d2ebfc482f26bb14847098e5a9097185d973","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-29 13:59:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82e3158aafb4e398c8901d45f58d560140e83999","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-29 22:12:27.000000000","message":"Patch Set 1: Code-Review+1\n\nYes please","accounts_in_message":[],"_revision_number":1},{"id":"772e491ac88db2dc1caea7879ab68e75fd149630","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-30 01:21:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"170baa24d05f3e26b4a99d01b432684e53c843f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 01:21:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f01cdb32231bf4f9e2ed181c27f744ea6cdf416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 01:34:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/26/661926/1/gate/openstack-tox-docs/8a28a14/html/ : SUCCESS in 7m 44s\n- openstack-tox-validate http://logs.openstack.org/26/661926/1/gate/openstack-tox-validate/3509606/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0d6d399ac9a52b5f46f932198e0ff96292e92656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 01:34:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92e8e4373c8cfb8c9d2dbbbeed7157aba1b2e1e9","revisions":{"92e8e4373c8cfb8c9d2dbbbeed7157aba1b2e1e9":{"kind":"REWORK","_number":1,"created":"2019-05-29 09:48:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/661926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/661926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/661926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/661926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/661926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/661926/1"}}},"commit":{"parents":[{"commit":"19e1a0990f61c409a50e9aa6a6d4998b4ee7b6e9","subject":"Merge \"Release Horizon Stein 15.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19e1a0990f61c409a50e9aa6a6d4998b4ee7b6e9"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-29 09:44:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-29 09:48:11.000000000","tz":60},"subject":"Release openstackdocstheme 1.30.0","message":"Release openstackdocstheme 1.30.0\n\nSome minor updates including fixing issues with the HTML5 output\ngenerated by Sphinx 2.0, inlining the OpenStack logo rather than\nreferencing it from elsewhere, and adding a new SOURCE_DATE_EPOCH option\nfor use in reproducible builds.\n\nChange-Id: Ie6743fa52434abf8ba0f0aa4db741552e6ccfb95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92e8e4373c8cfb8c9d2dbbbeed7157aba1b2e1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92e8e4373c8cfb8c9d2dbbbeed7157aba1b2e1e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
