)]}'
{"id":"openstack%2Freleases~662606","triplet_id":"openstack%2Freleases~master~Ie5634e1ec36cee418770c51b4b8c1efe40e3f104","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ie5634e1ec36cee418770c51b4b8c1efe40e3f104","subject":"Release 0.29.0 of openstacksdk","status":"MERGED","created":"2019-06-01 13:25:06.000000000","updated":"2019-06-01 14:23:54.000000000","submitted":"2019-06-01 14:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662606-1559399034109-0e238d65","meta_rev_id":"27bb9c79cb6ede7cff62a6ce1c0a0e5106fe6b18","_number":662606,"virtual_id_number":662606,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-01 14:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-01 14:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-01 14:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-01 14:14:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-01 14:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"532708678fcaab10e951458d4f766185d8bee7ca","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-06-01 13:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44d2f5e54060fee94160fb24957fb0f476beb53c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 13:36:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/662606/1/check/openstack-tox-docs/f5c07b0/html/ : SUCCESS in 6m 38s\n- openstack-tox-validate http://logs.openstack.org/06/662606/1/check/openstack-tox-validate/29f39e1/ : SUCCESS in 3m 48s\n- releases-tox-list-changes http://logs.openstack.org/06/662606/1/check/releases-tox-list-changes/b6246f6/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a03d65f5325785f92353808323fefaa9e47011a2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-01 14:14:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe5f621883f215770ba0ab429dd3795547178173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 14:14:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e18b3764a17dc07334cf5399d21a842eb7bfe52c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 14:23:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/662606/1/gate/openstack-tox-docs/2095230/html/ : SUCCESS in 5m 35s\n- openstack-tox-validate http://logs.openstack.org/06/662606/1/gate/openstack-tox-validate/54a072c/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"27bb9c79cb6ede7cff62a6ce1c0a0e5106fe6b18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 14:23:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2aff12021b06dd172a7feaf304c416de20121042","revisions":{"2aff12021b06dd172a7feaf304c416de20121042":{"kind":"REWORK","_number":1,"created":"2019-06-01 13:25:06.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/06/662606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/662606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/662606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/662606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/662606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/662606/1"}}},"commit":{"parents":[{"commit":"49e5e77144895c57cdf77ab56116ad5d7ff4f700","subject":"Merge \"Oslo Queens releases for 2019-05-29\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/49e5e77144895c57cdf77ab56116ad5d7ff4f700"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-06-01 13:24:42.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-06-01 13:25:01.000000000","tz":-300},"subject":"Release 0.29.0 of openstacksdk","message":"Release 0.29.0 of openstacksdk\n\nChange-Id: Ie5634e1ec36cee418770c51b4b8c1efe40e3f104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2aff12021b06dd172a7feaf304c416de20121042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2aff12021b06dd172a7feaf304c416de20121042"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
