)]}'
{"id":"openstack%2Freleases~662776","triplet_id":"openstack%2Freleases~master~Ia89cef5b54af406d89f5436198320d06f54ab461","project":"openstack/releases","branch":"master","topic":"bug/1831424","hashtags":[],"change_id":"Ia89cef5b54af406d89f5436198320d06f54ab461","subject":"Revert \"Replace the invalid index of nova highlight in Rocky\"","status":"MERGED","created":"2019-06-03 14:42:57.000000000","updated":"2019-06-04 01:00:22.000000000","submitted":"2019-06-03 22:17:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662776-1559600235617-caf50bd9","meta_rev_id":"f7514d59d998a4ec6f5d5a9cf6cbc3286408797c","_number":662776,"virtual_id_number":662776,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-06-03 22:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-03 22:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-03 21:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-03 22:04:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 14:42:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-03 14:42:57.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-06-03 21:58:24.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-06-03 22:04:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-03 22:17:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd1a3a834be3c10f558be9b900a5f490b85934b2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-03 14:42:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8d5475f29eeb868e792de6bda472b489ecb61c6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-03 14:43:19.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8436a3b0a79f910b8baecc9fed8b3e578c958cdf","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-03 14:43:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9507f4ef94e5c8b95f4e0bcb5139392d52ca8229","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:06:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/662776/2/check/openstack-tox-docs/16993bd/html/ : SUCCESS in 8m 03s\n- openstack-tox-validate http://logs.openstack.org/76/662776/2/check/openstack-tox-validate/4c629ee/ : SUCCESS in 4m 39s\n- releases-tox-list-changes http://logs.openstack.org/76/662776/2/check/releases-tox-list-changes/9ca341a/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"6d72399315660ca45e7f3f8e3d247de65fa8e0df","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-03 21:58:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bf5e0f222049e52a6212a0a93d0da911c53adf1a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 22:04:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e60641ab38b1305cd20d7fbcb87fc6f95e28339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 22:04:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5108dca0cdb234cb97cb566969faa125764eeccc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 22:17:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/662776/2/gate/openstack-tox-docs/8174a1d/html/ : SUCCESS in 7m 12s\n- openstack-tox-validate http://logs.openstack.org/76/662776/2/gate/openstack-tox-validate/a789777/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"99c6756ec3d6ed2c540fe709be8c86956ff63fa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 22:17:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"54d67b43e0c5bbd35b91f7d0e4f4f4b2f5eaa8f1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-04 01:00:22.000000000","message":"Patch Set 2:\n\nLGTM","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8eba75c9f22367f333e6e0a967c8b70d6b98584","revisions":{"e3c148067e0c964c3f886ae3f1b1565215026c4b":{"kind":"REWORK","_number":1,"created":"2019-06-03 14:42:57.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/76/662776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/662776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/662776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/662776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/662776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/662776/1"}}},"commit":{"parents":[{"commit":"7b834e1772440fb56e13600f81c2d729cfb181f6","subject":"Replace the invalid index of nova highlight in Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b834e1772440fb56e13600f81c2d729cfb181f6"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-03 14:42:57.000000000","tz":0},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-03 14:42:57.000000000","tz":0},"subject":"Revert \"Replace the invalid index of nova highlight in Rocky\"","message":"Revert \"Replace the invalid index of nova highlight in Rocky\"\n\nThis reverts commit 7b834e1772440fb56e13600f81c2d729cfb181f6.\n\nRather than point at the extracted placement policy docs\non \"latest\" this makes the rocky highlights point at the\nrocky-specific docs when placement was still in nova.\n\nChange-Id: Ia89cef5b54af406d89f5436198320d06f54ab461\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e3c148067e0c964c3f886ae3f1b1565215026c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e3c148067e0c964c3f886ae3f1b1565215026c4b"}]},"branch":"refs/heads/master"},"b8eba75c9f22367f333e6e0a967c8b70d6b98584":{"kind":"REWORK","_number":2,"created":"2019-06-03 14:43:32.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/76/662776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/662776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/662776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/662776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/662776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/662776/2"}}},"commit":{"parents":[{"commit":"7b834e1772440fb56e13600f81c2d729cfb181f6","subject":"Replace the invalid index of nova highlight in Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b834e1772440fb56e13600f81c2d729cfb181f6"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-03 14:42:57.000000000","tz":0},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-03 14:43:30.000000000","tz":0},"subject":"Revert \"Replace the invalid index of nova highlight in Rocky\"","message":"Revert \"Replace the invalid index of nova highlight in Rocky\"\n\nThis reverts commit 7b834e1772440fb56e13600f81c2d729cfb181f6.\n\nRather than point at the extracted placement policy docs\non \"latest\" this makes the rocky highlights point at the\nrocky-specific docs when placement was still in nova.\n\nChange-Id: Ia89cef5b54af406d89f5436198320d06f54ab461\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8eba75c9f22367f333e6e0a967c8b70d6b98584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8eba75c9f22367f333e6e0a967c8b70d6b98584"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
