)]}'
{"id":"openstack%2Freleases~662807","triplet_id":"openstack%2Freleases~master~Ie7e5c8c5c456a44ee750b5371cfc01a6763f4c21","project":"openstack/releases","branch":"master","topic":"release_osa","hashtags":[],"change_id":"Ie7e5c8c5c456a44ee750b5371cfc01a6763f4c21","subject":"Release OpenStack-Ansible Queens","status":"MERGED","created":"2019-06-03 17:27:30.000000000","updated":"2019-06-06 06:23:41.000000000","submitted":"2019-06-06 06:23:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662807-1559802221083-589de8e3","meta_rev_id":"ac78776a4f27a30d318b531eb096fc021411e4f0","_number":662807,"virtual_id_number":662807,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-06 06:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-06-06 06:09:17.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-04 14:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-05 17:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-05 10:52:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-06-06 06:09:17.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 14:33:38.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-05 10:52:34.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-06-05 17:27:02.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-06 06:09:17.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-06 06:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4676e3823ddcbe5d28586707c072bb939911514","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-03 17:27:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05afc701bfcf44172fc73c77b1907aaa63a5df45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 17:42:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/07/662807/1/check/openstack-tox-docs/6245ad4/html/ : SUCCESS in 8m 27s\n- openstack-tox-validate http://logs.openstack.org/07/662807/1/check/openstack-tox-validate/fd8e1ee/ : SUCCESS in 6m 58s\n- releases-tox-list-changes http://logs.openstack.org/07/662807/1/check/releases-tox-list-changes/05be59e/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ef0147daf51c2284c70c3a3ce92f252e23cad9c3","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-04 14:33:34.000000000","message":"Patch Set 1:\n\nptl+1","accounts_in_message":[],"_revision_number":1},{"id":"cbb98b1e407c7ff13f48b57678b7bf3edc99fdff","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-04 14:33:38.000000000","message":"Patch Set 1: Code-Review+1\n\nptl+1","accounts_in_message":[],"_revision_number":1},{"id":"9e7cf38b7ef49fb33f9e70be584a2544f8f4b6da","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-04 14:43:12.000000000","message":"Patch Set 1: Code-Review+1\n\n(with osa hat on:) Ok on the sha\n(with releases hat on:) will upgrade my vote when I know I am doing my usual releases reviews","accounts_in_message":[],"_revision_number":1},{"id":"c80ef1d19849d26f294daece0e7fe50ab13f03ad","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-06-05 10:52:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d696a6026f993000765b607e1c9248417d0f3de6","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-05 17:27:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"602cf4682e932de15e7224ad3640cbad90255fcf","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-06 06:09:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f55ca84ce14619c4f58cdc983a8186961e3dcaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:09:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bb31014b2ddc86d68822dbcf346a5d50b7dd902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:23:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/07/662807/1/gate/openstack-tox-docs/91c5de5/html/ : SUCCESS in 7m 34s\n- openstack-tox-validate http://logs.openstack.org/07/662807/1/gate/openstack-tox-validate/703e21b/ : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7fd2e50670c17dcba29bc0f039127f9d2d70fbbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:23:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d163efb13c2a2343aa36809e1a7821d1216e6109","revisions":{"d163efb13c2a2343aa36809e1a7821d1216e6109":{"kind":"REWORK","_number":1,"created":"2019-06-03 17:27:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/07/662807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/662807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/662807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/662807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/662807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/662807/1"}}},"commit":{"parents":[{"commit":"191ca30b9593e80c8ab47f99c2dc5604768c03fb","subject":"Merge \"Release 0.13.0 of os-traits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/191ca30b9593e80c8ab47f99c2dc5604768c03fb"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-06-03 17:27:06.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-06-03 17:27:06.000000000","tz":180},"subject":"Release OpenStack-Ansible Queens","message":"Release OpenStack-Ansible Queens\n\nChange-Id: Ie7e5c8c5c456a44ee750b5371cfc01a6763f4c21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d163efb13c2a2343aa36809e1a7821d1216e6109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d163efb13c2a2343aa36809e1a7821d1216e6109"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
