)]}'
{"id":"openstack%2Freleases~662955","triplet_id":"openstack%2Freleases~master~I1b1c444c5533ce0ada420c6c573993da72892c79","project":"openstack/releases","branch":"master","topic":"bifrost","hashtags":[],"change_id":"I1b1c444c5533ce0ada420c6c573993da72892c79","subject":"Release bifrost 6.1.0 for Train","status":"MERGED","created":"2019-06-04 07:26:59.000000000","updated":"2019-06-06 06:02:21.000000000","submitted":"2019-06-06 06:02:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662955-1559800941272-f390cf89","meta_rev_id":"09d1c2c899ff8d5d0bb2db568e4856a79c634561","_number":662955,"virtual_id_number":662955,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-06-06 06:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-06-06 05:53:59.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-05 15:26:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-06-06 05:53:59.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 15:26:38.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-06-06 05:53:59.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-06 06:02:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e896a6689bcc9cd7747f8560ab8fe9f9d052b2c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-04 07:26:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eee475f6a215762ecaa0be923a2c6121cb4c06e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 08:22:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/662955/1/check/openstack-tox-docs/8548ae4/html/ : SUCCESS in 7m 35s\n- openstack-tox-validate http://logs.openstack.org/55/662955/1/check/openstack-tox-validate/85e6ce5/ : RETRY_LIMIT in 1m 42s\n- releases-tox-list-changes http://logs.openstack.org/55/662955/1/check/releases-tox-list-changes/e9ad1ec/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e36eafe907d1de24c93c9ba2526c5b4b309887b1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-04 09:38:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d32610a3926d5d96e86b69971efefd3662362103","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 10:10:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/662955/1/check/openstack-tox-docs/821b147/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/55/662955/1/check/openstack-tox-validate/8ced1c0/ : SUCCESS in 4m 08s\n- releases-tox-list-changes http://logs.openstack.org/55/662955/1/check/releases-tox-list-changes/1a8999d/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ad399867a8494c8ecc7a91160974594ba67728cf","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-04 12:23:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3757c9dd4148614796886a9073246fd8f95be615","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 12:51:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/662955/2/check/openstack-tox-docs/f358d96/html/ : SUCCESS in 6m 26s\n- openstack-tox-validate http://logs.openstack.org/55/662955/2/check/openstack-tox-validate/2d1528e/ : SUCCESS in 4m 16s\n- releases-tox-list-changes http://logs.openstack.org/55/662955/2/check/releases-tox-list-changes/abd705c/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"240aa157c4c90792682b39f941aa8e5cdb1d0368","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-06-05 15:26:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"34c81efb2624d91143b75942905bab7d16a8cc34","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-06 05:53:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eccac9758df9581ee7c5c56659e0fb90935b861e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 05:54:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65a29a950561bbbd396b0dad553364221dac217e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:02:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/662955/2/gate/openstack-tox-docs/eb9e262/html/ : SUCCESS in 7m 01s\n- openstack-tox-validate http://logs.openstack.org/55/662955/2/gate/openstack-tox-validate/d25c085/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cc76ba07a86c3caefe7216f2f6171f962cf778f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:02:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13ae0b70bfaac859a258e31216d653803b591963","revisions":{"27f1e28f82b35b3f0231103866c037101f7aec07":{"kind":"REWORK","_number":1,"created":"2019-06-04 07:26:59.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/55/662955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/662955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/662955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/662955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/662955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/662955/1"}}},"commit":{"parents":[{"commit":"1bb20511b4375b1ab57baf6bebf5403a679cc829","subject":"Merge \"Revert \"Replace the invalid index of nova highlight in Rocky\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bb20511b4375b1ab57baf6bebf5403a679cc829"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-04 07:25:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-04 07:25:42.000000000","tz":120},"subject":"Release bifrost 6.1.0 for Train","message":"Release bifrost 6.1.0 for Train\n\nChange-Id: I1b1c444c5533ce0ada420c6c573993da72892c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/27f1e28f82b35b3f0231103866c037101f7aec07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/27f1e28f82b35b3f0231103866c037101f7aec07"}]},"branch":"refs/heads/master"},"13ae0b70bfaac859a258e31216d653803b591963":{"kind":"REWORK","_number":2,"created":"2019-06-04 12:23:59.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/55/662955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/662955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/662955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/662955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/662955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/662955/2"}}},"commit":{"parents":[{"commit":"1bb20511b4375b1ab57baf6bebf5403a679cc829","subject":"Merge \"Revert \"Replace the invalid index of nova highlight in Rocky\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bb20511b4375b1ab57baf6bebf5403a679cc829"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-04 07:25:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-04 12:23:46.000000000","tz":120},"subject":"Release bifrost 6.1.0 for Train","message":"Release bifrost 6.1.0 for Train\n\nChange-Id: I1b1c444c5533ce0ada420c6c573993da72892c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13ae0b70bfaac859a258e31216d653803b591963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13ae0b70bfaac859a258e31216d653803b591963"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
