)]}'
{"id":"openstack%2Freleases~663580","triplet_id":"openstack%2Freleases~master~I2879de0df530ccf96ac4be61c42ce07472399b4e","project":"openstack/releases","branch":"master","topic":"add-tag-jobs","hashtags":[],"change_id":"I2879de0df530ccf96ac4be61c42ce07472399b4e","subject":"Document tag pipeline jobs in contributor doc","status":"MERGED","created":"2019-06-06 10:12:03.000000000","updated":"2019-06-14 00:55:34.000000000","submitted":"2019-06-14 00:55:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663580-1560473734908-b1f8f8f4","meta_rev_id":"1ede223e4c3d8df8c41eb0ca9c71154ab3f2c3dc","_number":663580,"virtual_id_number":663580,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-06-14 00:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-06 11:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-14 00:45:00.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-14 00:45:00.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 11:00:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-14 00:45:00.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-14 00:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4d9d7b7b810f8aa83ae81f38f2be5b469a18629","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-06 10:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df57c2a9203adbe4b11514c6ad954d7f3d59b0f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:22:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/80/663580/1/check/openstack-tox-py35/da498be/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/80/663580/1/check/openstack-tox-py36/c392ae4/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/80/663580/1/check/openstack-tox-docs/2fe456d/html/ : SUCCESS in 5m 57s\n- openstack-tox-validate http://logs.openstack.org/80/663580/1/check/openstack-tox-validate/fbbfe91/ : SUCCESS in 5m 09s\n- releases-tox-list-changes http://logs.openstack.org/80/663580/1/check/releases-tox-list-changes/7ba5bad/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/80/663580/1/check/openstack-tox-pep8/a18c82e/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b718ed519dbc49f7a610e1d70883258d7335adb9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-06 11:00:07.000000000","message":"Patch Set 1: Code-Review+2\n\nGreat!","accounts_in_message":[],"_revision_number":1},{"id":"59b7f114cdd314f0b8eb609f40127fcdd323649b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-14 00:45:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d637568954a33de01687234dbcf0776c6d88e2fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 00:45:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"147c43d3119a48ca61d4a35db98553368154c65e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 00:55:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ab84740e22fba5f95af2b61d71bb64b31bf9e855","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 00:55:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/80/663580/1/gate/openstack-tox-py35/4f1fbf1/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/80/663580/1/gate/openstack-tox-py36/b43e1a9/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/80/663580/1/gate/openstack-tox-docs/b6b97a6/html/ : SUCCESS in 8m 26s\n- openstack-tox-validate http://logs.openstack.org/80/663580/1/gate/openstack-tox-validate/c2aa13e/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f0dc28878715eb4ab8d5f71c6c3ed0a3eb64540","revisions":{"6f0dc28878715eb4ab8d5f71c6c3ed0a3eb64540":{"kind":"REWORK","_number":1,"created":"2019-06-06 10:12:03.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/80/663580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/663580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/663580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/663580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/663580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/663580/1"}}},"commit":{"parents":[{"commit":"a608cd7ec534c898d073f3fbdf6be499c3cdd889","subject":"Merge \"Release ironic 10.1.9 for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a608cd7ec534c898d073f3fbdf6be499c3cdd889"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-06 10:10:45.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-06 10:10:45.000000000","tz":120},"subject":"Document tag pipeline jobs in contributor doc","message":"Document tag pipeline jobs in contributor doc\n\nWhile debugging a release failure this morning I realized that\nthe magic job diagram was missing the release notes publication\njobs that are in the \"tag\" pipeline.\n\nChange-Id: I2879de0df530ccf96ac4be61c42ce07472399b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f0dc28878715eb4ab8d5f71c6c3ed0a3eb64540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f0dc28878715eb4ab8d5f71c6c3ed0a3eb64540"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
