)]}'
{"id":"openstack%2Freleases~663643","triplet_id":"openstack%2Freleases~master~Icca2038e8370ec9a60c367b4a36ac6bf589522e9","project":"openstack/releases","branch":"master","topic":"wci-m1-releases","hashtags":[],"change_id":"Icca2038e8370ec9a60c367b4a36ac6bf589522e9","subject":"[winstackers] Create milestone-1 releases","status":"MERGED","created":"2019-06-06 13:09:12.000000000","updated":"2019-06-07 11:41:51.000000000","submitted":"2019-06-07 11:41:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663643-1559907711100-589de8e3","meta_rev_id":"6b7dac17a813bf981e46377c8b2c48ad04b7430c","_number":663643,"virtual_id_number":663643,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":2,"date":"2019-06-07 11:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-06 18:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-07 10:29:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-07 11:29:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 10:29:42.000000000","updated_by":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"reviewer":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"state":"REVIEWER"},{"updated":"2019-06-07 11:29:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-07 11:41:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9f2ae8a2eec119944c6955bbab381fa7a928c01","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-06 13:09:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c4860b66a1860e398b97bef320dd765bdb47b59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/43/663643/1/check/openstack-tox-docs/b7d2a3b/html/ : SUCCESS in 6m 58s\n- openstack-tox-validate http://logs.openstack.org/43/663643/1/check/openstack-tox-validate/b33677a/ : SUCCESS in 4m 09s\n- releases-tox-list-changes http://logs.openstack.org/43/663643/1/check/releases-tox-list-changes/48f1e48/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"00084ba72daa91205c3649f990198fa417c075a6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-06 18:06:58.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, needs PTL ack.","accounts_in_message":[],"_revision_number":1},{"id":"50855599c448401168f52d661cdad3066b656d6e","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2019-06-07 10:29:42.000000000","message":"Patch Set 1: Code-Review+1\n\nTag, commit id looks good.","accounts_in_message":[],"_revision_number":1},{"id":"bf7baba7a3fc2eff13f3f06f054968b6fc963c3d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-07 11:29:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa3b97bfabaab8ae0f650bcfcbdb8e07c6be7b11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:30:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb0fc1983c5d1d79bb940aa1b44f107e449cbf03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:41:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/43/663643/1/gate/openstack-tox-docs/3cba94a/html/ : SUCCESS in 6m 44s\n- openstack-tox-validate http://logs.openstack.org/43/663643/1/gate/openstack-tox-validate/2cd98cb/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bb8eabb0c64c29dd838563a4b4093789f858c4f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:41:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ea1c811b8eda89304981bae169cc7f704fe962f","revisions":{"1ea1c811b8eda89304981bae169cc7f704fe962f":{"kind":"REWORK","_number":1,"created":"2019-06-06 13:09:12.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/43/663643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/663643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/663643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/663643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/663643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/663643/1"}}},"commit":{"parents":[{"commit":"a608cd7ec534c898d073f3fbdf6be499c3cdd889","subject":"Merge \"Release ironic 10.1.9 for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a608cd7ec534c898d073f3fbdf6be499c3cdd889"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-06-06 13:07:07.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-06-06 13:07:07.000000000","tz":600},"subject":"[winstackers] Create milestone-1 releases","message":"[winstackers] Create milestone-1 releases\n\nThis is a procedural review to create new releases for libraries following cycle-with-intermediary  model with code changes\n\nTeams should update this change as appropriate\n\nChange-Id: Icca2038e8370ec9a60c367b4a36ac6bf589522e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1ea1c811b8eda89304981bae169cc7f704fe962f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1ea1c811b8eda89304981bae169cc7f704fe962f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
