)]}'
{"id":"openstack%2Freleases~663670","triplet_id":"openstack%2Freleases~master~If8a2c2d5160f2d75ae2d67bf662f8735851b99c0","project":"openstack/releases","branch":"master","topic":"tripleo/train/m1","hashtags":[],"change_id":"If8a2c2d5160f2d75ae2d67bf662f8735851b99c0","subject":"TripleO train m1 release","status":"MERGED","created":"2019-06-06 13:42:36.000000000","updated":"2019-06-07 19:24:56.000000000","submitted":"2019-06-07 19:24:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663670-1559935496252-26b96e4b","meta_rev_id":"52bb47c1226c061555e421a241f168433e61ea6c","_number":663670,"virtual_id_number":663670,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-07 19:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-07 18:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-07 19:11:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 13:58:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-07 19:11:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-07 19:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5c010e285b2446aafcd119b4defe2461959e787","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 13:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dba82660dd3821924985451e723f28a9aabbde34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:55:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/663670/1/check/openstack-tox-docs/04c7624/html/ : SUCCESS in 7m 54s\n- openstack-tox-validate http://logs.openstack.org/70/663670/1/check/openstack-tox-validate/b1a2231/ : SUCCESS in 5m 58s\n- releases-tox-list-changes http://logs.openstack.org/70/663670/1/check/releases-tox-list-changes/22ba4cf/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d8b9dc53025c3bdc59081d627671326653343e05","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 13:58:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d0c0a026ea3692f15a4c031c14281f63a36fd96a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-06 18:11:22.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nSome verification of intent needed first.","accounts_in_message":[],"_revision_number":1},{"id":"29cd22ef325ab5d5edb9a5c8cb92d97b0d071fd9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-07 18:39:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f2e7f911037ee073b05069269ffb8d78546fe79b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-07 18:45:28.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks Emilien! Just waiting for CI results.","accounts_in_message":[],"_revision_number":2},{"id":"6ae3a79602fe8230313f319cc0acced4d036adbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:08:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/663670/2/check/openstack-tox-docs/9549b8a/html/ : SUCCESS in 6m 39s\n- openstack-tox-validate http://logs.openstack.org/70/663670/2/check/openstack-tox-validate/2a46b99/ : SUCCESS in 5m 44s\n- releases-tox-list-changes http://logs.openstack.org/70/663670/2/check/releases-tox-list-changes/c966fdb/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0c56ed8b6637959f57c7ff5bf162bd5f8993e457","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-07 19:11:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d23bd9524f84e8659c06135f7f358a0361beb84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:12:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f83934440d6a163864ef5011ff18e964615e716","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:24:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/663670/2/gate/openstack-tox-docs/c9b458a/html/ : SUCCESS in 7m 42s\n- openstack-tox-validate http://logs.openstack.org/70/663670/2/gate/openstack-tox-validate/e1871f8/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2378074b034eb8dcafd2e9e416442056a42d0526","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:24:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87b57236c10b428510f03eb8d547c2c86e649a7b","revisions":{"d0a24accb385e55171e07950921a7b0a908ef7c3":{"kind":"REWORK","_number":1,"created":"2019-06-06 13:42:36.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/70/663670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/663670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/663670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/663670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/663670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/663670/1"}}},"commit":{"parents":[{"commit":"a608cd7ec534c898d073f3fbdf6be499c3cdd889","subject":"Merge \"Release ironic 10.1.9 for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a608cd7ec534c898d073f3fbdf6be499c3cdd889"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 13:25:07.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 13:42:26.000000000","tz":-240},"subject":"TripleO train m1 release","message":"TripleO train m1 release\n\nChange-Id: If8a2c2d5160f2d75ae2d67bf662f8735851b99c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0a24accb385e55171e07950921a7b0a908ef7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0a24accb385e55171e07950921a7b0a908ef7c3"}]},"branch":"refs/heads/master"},"87b57236c10b428510f03eb8d547c2c86e649a7b":{"kind":"REWORK","_number":2,"created":"2019-06-07 18:39:18.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/70/663670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/663670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/663670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/663670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/663670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/663670/2"}}},"commit":{"parents":[{"commit":"a608cd7ec534c898d073f3fbdf6be499c3cdd889","subject":"Merge \"Release ironic 10.1.9 for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a608cd7ec534c898d073f3fbdf6be499c3cdd889"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 13:25:07.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 18:39:17.000000000","tz":0},"subject":"TripleO train m1 release","message":"TripleO train m1 release\n\nChange-Id: If8a2c2d5160f2d75ae2d67bf662f8735851b99c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87b57236c10b428510f03eb8d547c2c86e649a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87b57236c10b428510f03eb8d547c2c86e649a7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
