)]}'
{"id":"openstack%2Freleases~663738","triplet_id":"openstack%2Freleases~master~I9ac83bdcd5729985c3682594d54052b1861b6b3f","project":"openstack/releases","branch":"master","topic":"indepenedence","hashtags":[],"change_id":"I9ac83bdcd5729985c3682594d54052b1861b6b3f","subject":"Drop traits and resource-classes train placeholders","status":"MERGED","created":"2019-06-06 18:13:53.000000000","updated":"2019-06-07 09:03:55.000000000","submitted":"2019-06-07 03:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663738-1559877509833-71434b0a","meta_rev_id":"1aa6f154f484e522dc5ca94e851d7e7c89240918","_number":663738,"virtual_id_number":663738,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-07 03:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-06-07 03:03:38.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-06 18:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-06-07 03:03:38.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-06-06 18:14:02.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 18:14:02.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-06 18:15:23.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-07 03:03:38.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-07 03:18:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1445dcec347576cde017873d37040efdd38be30f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-06 18:13:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6b6fbc5ca745926b66a904647a91b35d4ea170a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-06 18:15:23.000000000","message":"Patch Set 1: Code-Review+1\n\nOkay, good, this makes sense.","accounts_in_message":[],"_revision_number":1},{"id":"bbe0aeb6c78b9b81f7675ac5ba8d0767b8b8b9ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 18:26:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/663738/1/check/openstack-tox-docs/fcdf1ec/html/ : SUCCESS in 6m 59s\n- openstack-tox-validate http://logs.openstack.org/38/663738/1/check/openstack-tox-validate/10090bb/ : SUCCESS in 3m 40s\n- releases-tox-list-changes http://logs.openstack.org/38/663738/1/check/releases-tox-list-changes/18972b5/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ccaf7d39b17289183b2d8eee6db2d9a63c22a010","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-07 03:03:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Sean","accounts_in_message":[],"_revision_number":1},{"id":"c5d6ede379b869bff03070fbe8fdc4cbe72b4280","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:03:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35ef75c29068da33de81eab5ae9f0dce9f810178","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:18:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4fd15337654d927aa1f9c1c29b4f898d13c0255c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:18:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/663738/1/gate/openstack-tox-docs/c10ec3f/html/ : SUCCESS in 8m 29s\n- openstack-tox-validate http://logs.openstack.org/38/663738/1/gate/openstack-tox-validate/ee9281f/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"43cba88779f6caa7a28bbbd91dc54c9642b92d53","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-07 09:03:55.000000000","message":"Patch Set 1:\n\nThanks for getting these. I wasn\u0027t sure what the right thing to do was when we made the switch to independent.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7f0d60e9fa31dd4af5fff006714257cc7af486c","revisions":{"d7f0d60e9fa31dd4af5fff006714257cc7af486c":{"kind":"REWORK","_number":1,"created":"2019-06-06 18:13:53.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/663738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/663738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/663738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/663738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/663738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/663738/1"}}},"commit":{"parents":[{"commit":"60d274489dcab68efd0a0aed7ce46684aec24d0f","subject":"Merge \"Release python-ironicclient 2.8.0 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60d274489dcab68efd0a0aed7ce46684aec24d0f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-06 18:13:15.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-06 18:13:15.000000000","tz":-300},"subject":"Drop traits and resource-classes train placeholders","message":"Drop traits and resource-classes train placeholders\n\nThese deliverables have moved to an independent release model starting\nin train.\n\nChange-Id: I9ac83bdcd5729985c3682594d54052b1861b6b3f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7f0d60e9fa31dd4af5fff006714257cc7af486c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7f0d60e9fa31dd4af5fff006714257cc7af486c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
