)]}'
{"id":"openstack%2Freleases~663773","triplet_id":"openstack%2Freleases~master~I2b2c3c76b039a9f084949ea065ea6f9163265cce","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I2b2c3c76b039a9f084949ea065ea6f9163265cce","subject":"Release python-openstackclient 3.19.0","status":"MERGED","created":"2019-06-06 20:52:50.000000000","updated":"2019-06-13 09:59:13.000000000","submitted":"2019-06-13 09:59:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663773-1560419953193-38ffb1f0","meta_rev_id":"9343b61270539d317e4f2b2fab2c42a59e4d84bf","_number":663773,"virtual_id_number":663773,"owner":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-06-07 03:23:38.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-06-10 19:26:14.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":2,"date":"2019-06-13 09:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-12 14:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-06-13 09:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-13 09:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 03:23:38.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-12 14:16:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-13 09:44:48.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-13 09:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb5a510243eff9009e7ebe81ce4abb96fba62222","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-06-06 20:52:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c94f9ab70f5eae4654733c7217e71ce8c8670cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:05:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/663773/1/check/openstack-tox-docs/5f7504b/html/ : SUCCESS in 10m 12s\n- openstack-tox-validate http://logs.openstack.org/73/663773/1/check/openstack-tox-validate/14d8108/ : SUCCESS in 4m 11s\n- releases-tox-list-changes http://logs.openstack.org/73/663773/1/check/releases-tox-list-changes/6d35bd4/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f05a60d89b56f7f3a0caf45c41df1b3932eb654f","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-06-06 21:18:08.000000000","message":"Patch Set 1: Workflow-1\n\nDang, missed one thing...please hold","accounts_in_message":[],"_revision_number":1},{"id":"738f3db1999db2927e98068fa13304f81583bca5","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-06-06 23:45:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7637db7df2506c48f484ff5333a9783f238b5aea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:00:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/663773/2/check/openstack-tox-docs/2940662/html/ : SUCCESS in 7m 45s\n- openstack-tox-validate http://logs.openstack.org/73/663773/2/check/openstack-tox-validate/2f47ad6/ : SUCCESS in 4m 24s\n- releases-tox-list-changes http://logs.openstack.org/73/663773/2/check/releases-tox-list-changes/85c84b2/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"43cfbf3161870b798a34f8fab825f132eac5849e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-07 03:23:38.000000000","message":"Patch Set 2:\n\nLooks ok to me.\n\nI have a question about I67e1dbc53cc0b37967c0011bcb2fc09bdef62d94 Are the removed modules part of the API?  should this be 4.0.0?","accounts_in_message":[],"_revision_number":2},{"id":"058a12aef67ab9722b9de11447e1a0c770f4ad6e","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-06-10 19:26:14.000000000","message":"Patch Set 2:\n\nThose are not part of the CLI (API), they are bits that moved to osc-lib a long time ago.  This is the last release of 3.x, we will be working on 4.0 now and removing those was cleanup I could safely do now.","accounts_in_message":[],"_revision_number":2},{"id":"6b07fdc4c283d2b5f8b16747a6c256eee8b1295b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-12 14:16:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3f769cff3e805c24440e39f3a2f3afc4aa08996e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-13 09:44:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4db2a81a4979d103644efc9c59cee17cdd398f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:45:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"91ed4a96e0dbbe7ea53329cb0f0fdc76bf73ac55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:59:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/663773/2/gate/openstack-tox-docs/28510d5/html/ : SUCCESS in 7m 54s\n- openstack-tox-validate http://logs.openstack.org/73/663773/2/gate/openstack-tox-validate/45c911f/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9343b61270539d317e4f2b2fab2c42a59e4d84bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:59:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7fa4181295b2e2c8a4d94bb93f736537f735cb15","revisions":{"c3fbae40329f492013ad52578be181e07b3fed26":{"kind":"REWORK","_number":1,"created":"2019-06-06 20:52:50.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/73/663773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/663773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/663773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/663773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/663773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/663773/1"}}},"commit":{"parents":[{"commit":"a608cd7ec534c898d073f3fbdf6be499c3cdd889","subject":"Merge \"Release ironic 10.1.9 for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a608cd7ec534c898d073f3fbdf6be499c3cdd889"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-06-06 20:52:26.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-06-06 20:52:26.000000000","tz":-300},"subject":"Release python-openstackclient 3.19.0","message":"Release python-openstackclient 3.19.0\n\nChange-Id: I2b2c3c76b039a9f084949ea065ea6f9163265cce\nSigned-off-by: Dean Troyer \u003cdtroyer@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3fbae40329f492013ad52578be181e07b3fed26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3fbae40329f492013ad52578be181e07b3fed26"}]},"branch":"refs/heads/master"},"7fa4181295b2e2c8a4d94bb93f736537f735cb15":{"kind":"REWORK","_number":2,"created":"2019-06-06 23:45:53.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/73/663773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/663773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/663773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/663773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/663773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/663773/2"}}},"commit":{"parents":[{"commit":"a608cd7ec534c898d073f3fbdf6be499c3cdd889","subject":"Merge \"Release ironic 10.1.9 for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a608cd7ec534c898d073f3fbdf6be499c3cdd889"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-06-06 20:52:26.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-06-06 23:45:43.000000000","tz":-300},"subject":"Release python-openstackclient 3.19.0","message":"Release python-openstackclient 3.19.0\n\nChange-Id: I2b2c3c76b039a9f084949ea065ea6f9163265cce\nSigned-off-by: Dean Troyer \u003cdtroyer@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7fa4181295b2e2c8a4d94bb93f736537f735cb15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7fa4181295b2e2c8a4d94bb93f736537f735cb15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
