)]}'
{"id":"openstack%2Freleases~665246","triplet_id":"openstack%2Freleases~master~Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d","project":"openstack/releases","branch":"master","topic":"liaisons_scripts","hashtags":[],"change_id":"Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d","subject":"Update Liaison Scripts","status":"MERGED","created":"2019-06-13 21:53:24.000000000","updated":"2019-06-20 08:37:54.000000000","submitted":"2019-06-20 08:37:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665246-1561019874162-7895b45e","meta_rev_id":"0e88f89dc0e9f18b30316c2a6ea32aa3375fcca6","_number":665246,"virtual_id_number":665246,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-06-20 08:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-20 00:10:08.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-06-20 08:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-20 08:15:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 23:56:25.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-06-14 18:05:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-20 00:10:08.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-20 08:15:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-20 08:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66589c161236afcfb375622226b6f038ff2edb0e","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-13 21:53:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"133bc9d8719b1f26fa3ea88ed49e1c8c4058ba2a","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-13 21:54:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2bd0978165aeca8b2c2b986038a6cdd7b0b378d4","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-13 21:54:35.000000000","message":"Patch Set 1:\n\nWould still love comments..","accounts_in_message":[],"_revision_number":1},{"id":"6bf84f06a5d591c0075e843172248ee7fd5c6129","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 22:21:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/46/665246/1/check/openstack-tox-py35/192ed26/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/46/665246/1/check/openstack-tox-py36/2f7d429/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/46/665246/1/check/openstack-tox-docs/6f9617c/html/ : SUCCESS in 8m 50s\n- openstack-tox-validate http://logs.openstack.org/46/665246/1/check/openstack-tox-validate/97f9e04/ : FAILURE in 4m 00s\n- releases-tox-list-changes http://logs.openstack.org/46/665246/1/check/releases-tox-list-changes/f345147/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/46/665246/1/check/openstack-tox-pep8/31d3bed/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bc35f0e7e5d54a2caacf642fa21d01eaa5d6e4aa","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-13 22:43:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4f116e830b1524881739cb90381fcee5149972a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 22:56:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/665246/2/check/openstack-tox-py35/2a604eb/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/46/665246/2/check/openstack-tox-py36/7439a3b/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/46/665246/2/check/openstack-tox-docs/8990ae3/html/ : SUCCESS in 7m 11s\n- openstack-tox-validate http://logs.openstack.org/46/665246/2/check/openstack-tox-validate/b0df1b3/ : SUCCESS in 3m 48s\n- releases-tox-list-changes http://logs.openstack.org/46/665246/2/check/releases-tox-list-changes/b9b8bf5/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/46/665246/2/check/openstack-tox-pep8/822034c/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1f2916a70afa046112ce85a22eed0b4f339423cc","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-13 23:56:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6308fd244bbe00ab71d0a5d0565e0c0b614a63c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-14 03:47:28.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nThanks.\n\nI think we need to update list-changes with the new data structure but looking good aprt from that.","accounts_in_message":[],"_revision_number":2},{"id":"c437a31f9aa11d386f265ffe0771deb0d84aae76","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-14 18:05:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9156e2521a5f0f58c2d413716ee999188f85d8c7","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-14 21:12:22.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"11adcb7e28af27fc20a936d6308bacd168fb4501","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-14 21:12:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c7f5bc142d78e1acbe3b4eeee2ce29f4ecec128","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-14 21:20:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fe3315fe36dc696678b25da6c78fd886be2315ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 21:34:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/46/665246/4/check/openstack-tox-py35/cea6b4e/ : SUCCESS in 6m 51s\n- openstack-tox-py36 http://logs.openstack.org/46/665246/4/check/openstack-tox-py36/2e37d0a/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/46/665246/4/check/openstack-tox-docs/4c916b0/html/ : SUCCESS in 7m 56s\n- openstack-tox-validate http://logs.openstack.org/46/665246/4/check/openstack-tox-validate/19ab20f/ : SUCCESS in 4m 17s\n- releases-tox-list-changes http://logs.openstack.org/46/665246/4/check/releases-tox-list-changes/7abc614/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/46/665246/4/check/openstack-tox-pep8/beb89a4/ : FAILURE in 3m 55s","accounts_in_message":[],"_revision_number":4},{"id":"f83f1b1981fbad54007d7b3f06074e7cbfd248a5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-17 09:03:32.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7e65e42a306e79ac1c91e3ade2199d5d52cc300a","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-18 20:12:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7f706f077c8b7de98898477657df69621a208a6a","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-18 20:12:41.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a08914916c3614bdee5a363789de472fc832a575","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:35:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/665246/5/check/openstack-tox-py35/7c2dd38/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/46/665246/5/check/openstack-tox-py36/14df8c6/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/46/665246/5/check/openstack-tox-docs/0fd4eab/html/ : SUCCESS in 7m 18s\n- openstack-tox-validate http://logs.openstack.org/46/665246/5/check/openstack-tox-validate/555dc01/ : SUCCESS in 4m 06s\n- releases-tox-list-changes http://logs.openstack.org/46/665246/5/check/releases-tox-list-changes/454e7f8/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/46/665246/5/check/openstack-tox-pep8/a959227/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c58c4136f27b2d7c3ab7f2e04f90a38424d66697","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-19 22:50:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4889537884bc1c1de31615b88d6ccb3579eadfea","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-19 23:22:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c8992e6f4b4a04858484e5910febee9f79c10ea7","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-19 23:22:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"94270b747c25469fee97d35ad4059c690f070c93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 23:52:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/665246/6/check/openstack-tox-py35/cdb30b0/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/46/665246/6/check/openstack-tox-py36/b514f2d/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/46/665246/6/check/openstack-tox-docs/091e305/html/ : SUCCESS in 7m 20s\n- openstack-tox-validate http://logs.openstack.org/46/665246/6/check/openstack-tox-validate/87af1b5/ : SUCCESS in 4m 26s\n- releases-tox-list-changes http://logs.openstack.org/46/665246/6/check/releases-tox-list-changes/1d33a73/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/46/665246/6/check/openstack-tox-pep8/11999a5/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":6},{"id":"60fce10ca55c2851b3f5086ac0d0f286864358e6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-20 00:10:08.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ea6eb8dea57aa920b7ce18d49f3a0fc4316cb6f8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-20 08:15:57.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a98ac097e44d612d409e059ce6923222b0116a8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 08:16:09.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7e641f36bdcf2d642449ab83a28bfb2a351d0419","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 08:37:53.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/665246/6/gate/openstack-tox-py35/ef65c9e/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/46/665246/6/gate/openstack-tox-py36/c1a8a92/ : SUCCESS in 3m 59s\n- openstack-tox-docs http://logs.openstack.org/46/665246/6/gate/openstack-tox-docs/e0feda4/html/ : SUCCESS in 7m 04s\n- openstack-tox-validate http://logs.openstack.org/46/665246/6/gate/openstack-tox-validate/25c5e51/ : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":6},{"id":"0ab9563657822781e554311f5ade79682811d73d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 08:37:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"59c38e67083797ea703ce3c7e81418681c820970","revisions":{"f25c898e4f4441892ae11cc870eafafa8a5d51fe":{"kind":"REWORK","_number":1,"created":"2019-06-13 21:53:24.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/46/665246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/665246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/665246/1"}}},"commit":{"parents":[{"commit":"6074d0cf6902372732335fc81df56b5d600c7c67","subject":"Merge \"Add Liaisons to Deliverables File\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6074d0cf6902372732335fc81df56b5d600c7c67"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-10 16:36:56.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-13 21:53:03.000000000","tz":-420},"subject":"Update Liaison Scripts","message":"Update Liaison Scripts\n\nNow that the release liaison info lives in data/release_liaisons.yaml,\nwe don\u0027t need to pull from the wiki. This patch also adds a schema\nto verify the release_liaions.yaml is formatted correctly.\n\nStory: 2005702\nTask: 33639\nTask: 33733\n\nChange-Id: Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f25c898e4f4441892ae11cc870eafafa8a5d51fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f25c898e4f4441892ae11cc870eafafa8a5d51fe"}]},"branch":"refs/heads/master"},"cd001bb198a00214d970dfe944eabe2d0ba912bb":{"kind":"REWORK","_number":2,"created":"2019-06-13 22:43:30.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/46/665246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/665246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/665246/2"}}},"commit":{"parents":[{"commit":"6074d0cf6902372732335fc81df56b5d600c7c67","subject":"Merge \"Add Liaisons to Deliverables File\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6074d0cf6902372732335fc81df56b5d600c7c67"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-10 16:36:56.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-13 22:42:51.000000000","tz":-420},"subject":"Update Liaison Scripts","message":"Update Liaison Scripts\n\nNow that the release liaison info lives in data/release_liaisons.yaml,\nwe don\u0027t need to pull from the wiki. This patch also adds a schema\nto verify the release_liaions.yaml is formatted correctly.\n\nStory: 2005702\nTask: 33639\nTask: 33733\n\nChange-Id: Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd001bb198a00214d970dfe944eabe2d0ba912bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd001bb198a00214d970dfe944eabe2d0ba912bb"}]},"branch":"refs/heads/master"},"ecd4427928b5dee49b18d80240e21269d1681fc7":{"kind":"REWORK","_number":3,"created":"2019-06-14 21:12:23.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/46/665246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/665246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/665246/3"}}},"commit":{"parents":[{"commit":"6074d0cf6902372732335fc81df56b5d600c7c67","subject":"Merge \"Add Liaisons to Deliverables File\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6074d0cf6902372732335fc81df56b5d600c7c67"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-10 16:36:56.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-14 21:12:12.000000000","tz":-420},"subject":"Update Liaison Scripts","message":"Update Liaison Scripts\n\nNow that the release liaison info lives in data/release_liaisons.yaml,\nwe don\u0027t need to pull from the wiki. This patch also adds a schema\nto verify the release_liaions.yaml is formatted correctly.\n\nStory: 2005702\nTask: 33639\nTask: 33733\n\nChange-Id: Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ecd4427928b5dee49b18d80240e21269d1681fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ecd4427928b5dee49b18d80240e21269d1681fc7"}]},"branch":"refs/heads/master"},"3c2fe85702c399c583e5f5e4602a981d5417bbe8":{"kind":"REWORK","_number":4,"created":"2019-06-14 21:20:36.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/46/665246/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/665246/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/665246/4"}}},"commit":{"parents":[{"commit":"6074d0cf6902372732335fc81df56b5d600c7c67","subject":"Merge \"Add Liaisons to Deliverables File\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6074d0cf6902372732335fc81df56b5d600c7c67"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-10 16:36:56.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-14 21:20:27.000000000","tz":-420},"subject":"Update Liaison Scripts","message":"Update Liaison Scripts\n\nNow that the release liaison info lives in data/release_liaisons.yaml,\nwe don\u0027t need to pull from the wiki. This patch also adds a schema\nto verify the release_liaions.yaml is formatted correctly.\n\nStory: 2005702\nTask: 33639\nTask: 33733\n\nChange-Id: Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c2fe85702c399c583e5f5e4602a981d5417bbe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c2fe85702c399c583e5f5e4602a981d5417bbe8"}]},"branch":"refs/heads/master"},"f250415afc413b0a478ebd239672fbf6cf05ac5a":{"kind":"REWORK","_number":5,"created":"2019-06-18 20:12:14.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/46/665246/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/665246/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/665246/5"}}},"commit":{"parents":[{"commit":"6074d0cf6902372732335fc81df56b5d600c7c67","subject":"Merge \"Add Liaisons to Deliverables File\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6074d0cf6902372732335fc81df56b5d600c7c67"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-10 16:36:56.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-18 20:10:47.000000000","tz":-420},"subject":"Update Liaison Scripts","message":"Update Liaison Scripts\n\nNow that the release liaison info lives in data/release_liaisons.yaml,\nwe don\u0027t need to pull from the wiki. This patch also adds a schema\nto verify the release_liaions.yaml is formatted correctly.\n\nStory: 2005702\nTask: 33639\nTask: 33733\n\nChange-Id: Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f250415afc413b0a478ebd239672fbf6cf05ac5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f250415afc413b0a478ebd239672fbf6cf05ac5a"}]},"branch":"refs/heads/master"},"59c38e67083797ea703ce3c7e81418681c820970":{"kind":"REWORK","_number":6,"created":"2019-06-19 23:22:37.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/46/665246/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/665246/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/665246/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/665246/6"}}},"commit":{"parents":[{"commit":"6074d0cf6902372732335fc81df56b5d600c7c67","subject":"Merge \"Add Liaisons to Deliverables File\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6074d0cf6902372732335fc81df56b5d600c7c67"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-10 16:36:56.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-06-19 23:22:29.000000000","tz":-420},"subject":"Update Liaison Scripts","message":"Update Liaison Scripts\n\nNow that the release liaison info lives in data/release_liaisons.yaml,\nwe don\u0027t need to pull from the wiki. This patch also adds a schema\nto verify the release_liaions.yaml is formatted correctly.\n\nStory: 2005702\nTask: 33639\nTask: 33733\n\nChange-Id: Ic4de16c26bb1a4fb878b86d2bd9c59bf5f54d11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59c38e67083797ea703ce3c7e81418681c820970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59c38e67083797ea703ce3c7e81418681c820970"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
