)]}'
{"id":"openstack%2Freleases~665300","triplet_id":"openstack%2Freleases~master~I04011d0929a511c32bb28de1ae8c68bb079f9137","project":"openstack/releases","branch":"master","topic":"glance_store/0.29.1","hashtags":[],"change_id":"I04011d0929a511c32bb28de1ae8c68bb079f9137","subject":"Release glance_store 0.29.1","status":"MERGED","created":"2019-06-14 05:40:37.000000000","updated":"2019-06-17 00:47:37.000000000","submitted":"2019-06-17 00:47:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665300-1560732457188-44d5107d","meta_rev_id":"52dd34d80581b7366a05d3772f7bc52d6b9fcdac","_number":665300,"virtual_id_number":665300,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-06-17 00:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-14 18:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-17 00:35:32.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-17 00:35:32.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 05:40:57.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-06-14 18:09:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-17 00:35:32.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-17 00:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd657080988feec9a45427d22840c10ced4df0e6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-14 05:40:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2657cc77b959d827ddb8f1979c358c32117e521","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 06:00:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/665300/1/check/openstack-tox-docs/86d204a/html/ : SUCCESS in 7m 17s\n- openstack-tox-validate http://logs.openstack.org/00/665300/1/check/openstack-tox-validate/7dc6cc0/ : FAILURE in 3m 49s\n- releases-tox-list-changes http://logs.openstack.org/00/665300/1/check/releases-tox-list-changes/8714314/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0f8167eeae5f12cc4036c89de91e3490ffcb97e1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-14 06:22:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5778aa105de0c4005d627df59b339fed7e3130e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:03:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/665300/2/check/openstack-tox-docs/df89d9e/html/ : SUCCESS in 6m 29s\n- openstack-tox-validate http://logs.openstack.org/00/665300/2/check/openstack-tox-validate/ef14366/ : SUCCESS in 4m 39s\n- releases-tox-list-changes http://logs.openstack.org/00/665300/2/check/releases-tox-list-changes/89de1d2/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"1ba10edebcbc15934e45f784ba7b7ddbe4532ba5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-14 18:09:03.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me. Just includes the revert of a backwards incompatible change and a new release note noting that there was an issue because of that, so should be good as a bugfix.\n\nWould be good to have Erno\u0027s ack on it.","accounts_in_message":[],"_revision_number":2},{"id":"e8c0bff2e1d2f36ffeb55958f67a2dbecb5c00ee","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-17 00:35:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3eb445deb679e5ef0d7e9feb4b47dc33e6becb1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 00:35:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c913ff9aecb2b72b3417c2f464a9ccab599f3469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 00:47:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/665300/2/gate/openstack-tox-docs/bd1ac3e/html/ : SUCCESS in 7m 00s\n- openstack-tox-validate http://logs.openstack.org/00/665300/2/gate/openstack-tox-validate/103035e/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"578c76f1a4f0d0c9d10c423e197eb26a15e9d78b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 00:47:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"01505dcbdb72a21e19f5691940f6f54d750853c9","revisions":{"b79b3ec07c3f18231dd92465438b23103431c6f5":{"kind":"REWORK","_number":1,"created":"2019-06-14 05:40:37.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/00/665300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/665300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/665300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/665300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/665300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/665300/1"}}},"commit":{"parents":[{"commit":"212b1c8a99033eefabe8e45122d93a56b2b8e80c","subject":"Merge \"Oslo releases for 2019-6-11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/212b1c8a99033eefabe8e45122d93a56b2b8e80c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-14 05:39:11.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-14 05:39:11.000000000","tz":0},"subject":"Release glance_store 0.29.1","message":"Release glance_store 0.29.1\n\nThis is the official Train milestone 1 release of glance_store.\n\nChange-Id: I04011d0929a511c32bb28de1ae8c68bb079f9137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b79b3ec07c3f18231dd92465438b23103431c6f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b79b3ec07c3f18231dd92465438b23103431c6f5"}]},"branch":"refs/heads/master"},"01505dcbdb72a21e19f5691940f6f54d750853c9":{"kind":"REWORK","_number":2,"created":"2019-06-14 06:22:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/00/665300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/665300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/665300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/665300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/665300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/665300/2"}}},"commit":{"parents":[{"commit":"212b1c8a99033eefabe8e45122d93a56b2b8e80c","subject":"Merge \"Oslo releases for 2019-6-11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/212b1c8a99033eefabe8e45122d93a56b2b8e80c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-14 05:39:11.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-14 06:21:21.000000000","tz":0},"subject":"Release glance_store 0.29.1","message":"Release glance_store 0.29.1\n\nThis is the official Train milestone 1 release of glance_store.\n\nChange-Id: I04011d0929a511c32bb28de1ae8c68bb079f9137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01505dcbdb72a21e19f5691940f6f54d750853c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01505dcbdb72a21e19f5691940f6f54d750853c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
