)]}'
{"id":"openstack%2Freleases~666546","triplet_id":"openstack%2Freleases~master~I83ff63cef18ae297013c3761a373078e580cf58b","project":"openstack/releases","branch":"master","topic":"cycle-automatic","hashtags":[],"change_id":"I83ff63cef18ae297013c3761a373078e580cf58b","subject":"Introduce cycle-automatic release model","status":"MERGED","created":"2019-06-20 11:51:25.000000000","updated":"2019-07-01 02:26:13.000000000","submitted":"2019-07-01 02:26:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666546-1561947973918-c2951fcc","meta_rev_id":"c5296d0cb8ffe11346526b141362082c6b85dcb9","_number":666546,"virtual_id_number":666546,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-07-01 02:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-20 15:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-01 02:06:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-21 08:46:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-01 02:06:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 14:16:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-21 08:46:18.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-07-01 02:06:53.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-01 02:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8625ce1888d35ac88d66da45019f139b412c082","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-20 11:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c76798572ff4ea34ddcf1470fd0dc1400c459cfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 12:07:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/666546/1/check/openstack-tox-py35/a255424/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/46/666546/1/check/openstack-tox-py36/56bdac5/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/46/666546/1/check/openstack-tox-docs/5a88e86/html/ : SUCCESS in 7m 27s\n- openstack-tox-validate http://logs.openstack.org/46/666546/1/check/openstack-tox-validate/b69ec8e/ : SUCCESS in 4m 16s\n- releases-tox-list-changes http://logs.openstack.org/46/666546/1/check/releases-tox-list-changes/ffa3076/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/46/666546/1/check/openstack-tox-pep8/24fbaf1/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"de42323000e100f88c04414ac5a629d056df6597","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-20 14:04:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"08f2fdf98af5baa7e31e8d39df26705d11f13959","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-20 14:16:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"93e9fc6c44f649b189e4312265b90d4c97a0bf53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 14:24:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/46/666546/2/check/openstack-tox-py35/f495399/ : SUCCESS in 5m 54s\n- openstack-tox-py36 http://logs.openstack.org/46/666546/2/check/openstack-tox-py36/0ce865b/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/46/666546/2/check/openstack-tox-docs/950c6fa/html/ : SUCCESS in 13m 31s\n- openstack-tox-validate http://logs.openstack.org/46/666546/2/check/openstack-tox-validate/b5cf23f/ : FAILURE in 3m 47s\n- releases-tox-list-changes http://logs.openstack.org/46/666546/2/check/releases-tox-list-changes/e9c31d3/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/46/666546/2/check/openstack-tox-pep8/f4bdd94/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2a3a346dd5b74e3a7dc56273b955b190685213a8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-20 15:54:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"90610f7986ae69918ecabd26038c1b6fafe88c12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:08:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/666546/3/check/openstack-tox-py35/10ffeba/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/46/666546/3/check/openstack-tox-py36/46e07cf/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/46/666546/3/check/openstack-tox-docs/b1587ca/html/ : SUCCESS in 7m 22s\n- openstack-tox-validate http://logs.openstack.org/46/666546/3/check/openstack-tox-validate/bce3d40/ : SUCCESS in 3m 48s\n- releases-tox-list-changes http://logs.openstack.org/46/666546/3/check/releases-tox-list-changes/76fe526/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/46/666546/3/check/openstack-tox-pep8/acc213a/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"d1e26943b8a4113ea9001e9ba67376c5a1a90859","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-06-21 08:46:18.000000000","message":"Patch Set 3: Code-Review+1\n\ntempest-plugins and similar has been problematic from OpenStack distros perspective (at least RDO). Automatic releases at RC looks like a good approach for them.","accounts_in_message":[],"_revision_number":3},{"id":"a2a4af6bbacf8e0339c1794f12ce413b305dda87","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-01 02:06:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThanks!\n\nLooks good to me.  I have a question and validation but that\u0027d be a separate thing anyway","accounts_in_message":[],"_revision_number":3},{"id":"4d0df92277a6fbf8e7c0be29899ab528853f6d0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 02:07:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"53086bee0520db31d2b93c517502a76469117c61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 02:26:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/666546/3/gate/openstack-tox-py35/8339e87/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/46/666546/3/gate/openstack-tox-py36/9a8883b/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/46/666546/3/gate/openstack-tox-docs/6eeebbb/html/ : SUCCESS in 6m 59s\n- openstack-tox-validate http://logs.openstack.org/46/666546/3/gate/openstack-tox-validate/29cea3a/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"cb43f3e555ba663419ec114a28af20f3831c9c2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 02:26:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b0f8d79dc07f60aab6832b47cb09a6ce128fbd64","revisions":{"2b97e7b248e7b440e69619cc71c06fdd67ce91d9":{"kind":"REWORK","_number":1,"created":"2019-06-20 11:51:25.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/46/666546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/666546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/666546/1"}}},"commit":{"parents":[{"commit":"a9e488f400bdabadefacb0835d54947b936c2577","subject":"Merge \"Release puppet-manila for Stein, Rocky and Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9e488f400bdabadefacb0835d54947b936c2577"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 10:24:54.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 10:46:02.000000000","tz":120},"subject":"Introduce cycle-automatic release model","message":"Introduce cycle-automatic release model\n\nFor things like tempest-plugins, the cycle-with-intermediary model\nis a bit overkill, especially now that we encourage multiple\nintermediary releases to qualify.\n\nIntroduce a new \u0027cycle-automatic\u0027 model for deliverables that only\nneed to be released once, automatically at the end of the cycle.\nThis is limited to \"other\" or \"tempest-plugin\" deliverables.\n\nTask: 31025\nChange-Id: I83ff63cef18ae297013c3761a373078e580cf58b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b97e7b248e7b440e69619cc71c06fdd67ce91d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b97e7b248e7b440e69619cc71c06fdd67ce91d9"}]},"branch":"refs/heads/master"},"d1444cd880a1ed7b60595e0d6100f7c092bc400a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-20 14:04:04.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/46/666546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/666546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/666546/2"}}},"commit":{"parents":[{"commit":"e6c30adbdebd9d4b24c7aa8c6bca6eea0674fc45","subject":"Fix regression in list-changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6c30adbdebd9d4b24c7aa8c6bca6eea0674fc45"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 10:24:54.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 14:04:04.000000000","tz":0},"subject":"Introduce cycle-automatic release model","message":"Introduce cycle-automatic release model\n\nFor things like tempest-plugins, the cycle-with-intermediary model\nis a bit overkill, especially now that we encourage multiple\nintermediary releases to qualify.\n\nIntroduce a new \u0027cycle-automatic\u0027 model for deliverables that only\nneed to be released once, automatically at the end of the cycle.\nThis is limited to \"other\" or \"tempest-plugin\" deliverables.\n\nTask: 31025\nChange-Id: I83ff63cef18ae297013c3761a373078e580cf58b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d1444cd880a1ed7b60595e0d6100f7c092bc400a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d1444cd880a1ed7b60595e0d6100f7c092bc400a"}]},"branch":"refs/heads/master"},"b0f8d79dc07f60aab6832b47cb09a6ce128fbd64":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-20 15:54:57.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/46/666546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/666546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/666546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/666546/3"}}},"commit":{"parents":[{"commit":"08cf70cbc6c8e170a342a7735f779156ac63dae2","subject":"Fix regression in list-changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/08cf70cbc6c8e170a342a7735f779156ac63dae2"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 10:24:54.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 15:54:57.000000000","tz":0},"subject":"Introduce cycle-automatic release model","message":"Introduce cycle-automatic release model\n\nFor things like tempest-plugins, the cycle-with-intermediary model\nis a bit overkill, especially now that we encourage multiple\nintermediary releases to qualify.\n\nIntroduce a new \u0027cycle-automatic\u0027 model for deliverables that only\nneed to be released once, automatically at the end of the cycle.\nThis is limited to \"other\" or \"tempest-plugin\" deliverables.\n\nTask: 31025\nChange-Id: I83ff63cef18ae297013c3761a373078e580cf58b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0f8d79dc07f60aab6832b47cb09a6ce128fbd64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0f8d79dc07f60aab6832b47cb09a6ce128fbd64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
