)]}'
{"id":"openstack%2Freleases~666562","triplet_id":"openstack%2Freleases~master~Id06794f165f0c852fb62dcacd53058bdc756899e","project":"openstack/releases","branch":"master","topic":"final-without-rc","hashtags":[],"change_id":"Id06794f165f0c852fb62dcacd53058bdc756899e","subject":"Do not allow final without RC in cycle-with-rc","status":"MERGED","created":"2019-06-20 12:59:49.000000000","updated":"2019-06-28 13:43:09.000000000","submitted":"2019-06-28 13:43:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666562-1561729389155-a45a0009","meta_rev_id":"7855ddc04ecbbcf4752418ebe842b8d08b6d3e61","_number":666562,"virtual_id_number":666562,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-06-28 13:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-28 13:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-20 23:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2019-06-24 18:45:53.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-28 13:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 23:12:13.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-06-24 18:45:53.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-06-28 13:29:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-28 13:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ffd1a2fe60d14359ed2c666efc53bd7170bca5d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-20 12:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbd5ff37acd99974592eb1a1d6ee8a64f3e160f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/62/666562/1/check/openstack-tox-py35/564efba/ : SUCCESS in 14m 32s\n- openstack-tox-py36 http://logs.openstack.org/62/666562/1/check/openstack-tox-py36/4038129/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/62/666562/1/check/openstack-tox-docs/42e266e/html/ : SUCCESS in 11m 51s\n- openstack-tox-validate http://logs.openstack.org/62/666562/1/check/openstack-tox-validate/085c7c1/ : SUCCESS in 4m 00s\n- releases-tox-list-changes http://logs.openstack.org/62/666562/1/check/releases-tox-list-changes/e05a436/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/62/666562/1/check/openstack-tox-pep8/2542561/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"65b04b9a6e70fc3cd88b3975171341a32766d9fc","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-06-20 23:12:13.000000000","message":"Patch Set 1: Code-Review+2\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":1},{"id":"a237f6f01458c4a853b699c7aef4cf29d2062e7c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-24 18:45:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"069f7c991091fa5e7dd880abde63e76b101dc1c1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-28 13:29:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66ee3fe65b84814a54a08409e484846e750dad74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:29:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7978e08186c1369ed4d56d353d5e9a9d501f13b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:43:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/62/666562/1/gate/openstack-tox-py35/fb4880a/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/62/666562/1/gate/openstack-tox-py36/e7af806/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/62/666562/1/gate/openstack-tox-docs/04d6d30/html/ : SUCCESS in 8m 03s\n- openstack-tox-validate http://logs.openstack.org/62/666562/1/gate/openstack-tox-validate/de8470d/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"36f9eef04ced11a60ab1d5adff659bc7bf969762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:43:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bcdeaaa9f90b4276b0d3a73bbc785e66814450b","revisions":{"1bcdeaaa9f90b4276b0d3a73bbc785e66814450b":{"kind":"REWORK","_number":1,"created":"2019-06-20 12:59:49.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/62/666562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/666562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/666562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/666562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/666562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/666562/1"}}},"commit":{"parents":[{"commit":"a9e488f400bdabadefacb0835d54947b936c2577","subject":"Merge \"Release puppet-manila for Stein, Rocky and Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9e488f400bdabadefacb0835d54947b936c2577"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 12:54:50.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 12:54:50.000000000","tz":120},"subject":"Do not allow final without RC in cycle-with-rc","message":"Do not allow final without RC in cycle-with-rc\n\nCurrently validation considers it valid to propose final versions\non a cycle-with-rc model without RC first:\n\nhttps://review.opendev.org/#/c/613444/2\n\nRelying on human review to catch that sounds risky. Let\u0027s fix it.\n\nChange-Id: Id06794f165f0c852fb62dcacd53058bdc756899e\nTask: 27762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bcdeaaa9f90b4276b0d3a73bbc785e66814450b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bcdeaaa9f90b4276b0d3a73bbc785e66814450b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
