)]}'
{"id":"openstack%2Freleases~666728","triplet_id":"openstack%2Freleases~master~I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56","project":"openstack/releases","branch":"master","topic":"release_osa","hashtags":[],"change_id":"I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56","subject":"Release OpenStack-Ansible Stein","status":"MERGED","created":"2019-06-20 20:19:06.000000000","updated":"2019-07-16 15:10:26.000000000","submitted":"2019-07-16 15:10:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666728-1563289826178-9977374c","meta_rev_id":"82f7fdd628c40b97563f319c104efa200161bacd","_number":666728,"virtual_id_number":666728,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-16 15:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-16 14:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-11 11:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-11 18:11:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2019-07-11 20:24:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-07-11 12:07:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 12:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-07-11 16:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-07-16 14:56:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 11:34:16.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-11 12:07:23.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-11 16:38:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-11 18:11:31.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-12 12:51:09.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-16 14:56:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-16 15:10:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f945c7f36b21f2d6c2734374a61eeff0bd9014d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-20 20:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa5748936034347f41913d08a7b9e98c48b988c6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-20 20:20:08.000000000","message":"Topic set to release_osa","accounts_in_message":[],"_revision_number":1},{"id":"66a1ccc3f5406b68f0edfe73d312bbe0f815e5f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:02:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666728/1/check/openstack-tox-docs/1938a5f/html/ : SUCCESS in 8m 43s\n- openstack-tox-validate http://logs.openstack.org/28/666728/1/check/openstack-tox-validate/98c059f/ : FAILURE in 5m 31s\n- releases-tox-list-changes http://logs.openstack.org/28/666728/1/check/releases-tox-list-changes/9c11615/ : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"466970be48d57ddcea6269181dd88765cda883b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-21 05:16:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9040dcda29608d468f025c2b542f425bcddd636d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 05:33:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666728/2/check/openstack-tox-docs/af0c9c9/html/ : SUCCESS in 7m 47s\n- openstack-tox-validate http://logs.openstack.org/28/666728/2/check/openstack-tox-validate/3ac7434/ : FAILURE in 4m 06s\n- releases-tox-list-changes http://logs.openstack.org/28/666728/2/check/releases-tox-list-changes/68f0f9d/ : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":2},{"id":"bae006ec969c40e9909fc379b774f3f291afea11","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-21 13:27:23.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"1d7a8abb644f00f845947f16a20b264795e578e2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 13:21:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9997f8280048bdd54609eed83af49d32f9efb9d8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 13:21:52.000000000","message":"Patch Set 3: Workflow-1\n\nI think we could be good there, but the latest keystone bugs makes me wonder if we shouldn\u0027t have a last tag.","accounts_in_message":[],"_revision_number":3},{"id":"3d20e652270bc8a401fe467b908e748a5ccf535e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:43:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/666728/3/check/openstack-tox-docs/2673678/html/ : SUCCESS in 7m 48s\n- openstack-tox-validate http://logs.openstack.org/28/666728/3/check/openstack-tox-validate/e5f8ca4/ : SUCCESS in 3m 57s\n- releases-tox-list-changes http://logs.openstack.org/28/666728/3/check/releases-tox-list-changes/3efd192/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"99ffdbac748c1124a2aad29aabbb6764fe9baa55","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-27 20:28:57.000000000","message":"Patch Set 3:\n\nI\u0027d vote for including keystone fix to the release w/o separate tag...","accounts_in_message":[],"_revision_number":3},{"id":"f8da60303981d39f3417a2b04fe73c902076b40b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 14:41:59.000000000","message":"Removed Workflow-1 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"5ecd57f327ee90633b8e0776f297bcd52fc13071","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 14:45:39.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"17be96e0e8055f655c09b025d66b552482b49375","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 14:46:23.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"d76103ac8884dfe7d27f1dc1ef25b1c7ae9d54f2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 14:48:34.000000000","message":"Patch Set 5:\n\nI would like to see mnaser\u0027s vote on the final one :)","accounts_in_message":[],"_revision_number":5},{"id":"a41273b564b383b37cfeca7934a72ada2bff02e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:01:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666728/5/check/openstack-tox-docs/93b8fd9/html/ : SUCCESS in 9m 03s\n- openstack-tox-validate http://logs.openstack.org/28/666728/5/check/openstack-tox-validate/112124a/ : FAILURE in 5m 30s\n- releases-tox-list-changes http://logs.openstack.org/28/666728/5/check/releases-tox-list-changes/cd39b20/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":5},{"id":"4884fc9a7b8ee766e4479ebee4a037472d4a7a93","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 15:02:33.000000000","message":"Patch Set 5:\n\nit looks like validations make it so we need to do an rc THEN release?","accounts_in_message":[],"_revision_number":5},{"id":"5b28ee47613407cc2dc4f99f5bb63e15870cb3e6","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 15:31:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"394dbb1e99b89630f07107d2764c38fb032219c2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 15:44:16.000000000","message":"Patch Set 6:\n\n\u003e it looks like validations make it so we need to do an rc THEN\n \u003e release?\n\nYeah, I thought we removed that, but no it\u0027s BAU :)","accounts_in_message":[],"_revision_number":6},{"id":"ce18af3acf7887a94208a44bc909fdab93d69b4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:44:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666728/6/check/openstack-tox-docs/78d8085/html/ : SUCCESS in 7m 09s\n- openstack-tox-validate http://logs.openstack.org/28/666728/6/check/openstack-tox-validate/28abe8d/ : FAILURE in 4m 37s\n- releases-tox-list-changes http://logs.openstack.org/28/666728/6/check/releases-tox-list-changes/ea84017/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":6},{"id":"087780df1acc7716386fa89c810becb31c34bdf2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 10:02:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d97253b1e96408e40ea6622d0032d8c3822a5cc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 10:16:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/666728/6/check/openstack-tox-docs/78b076d/html/ : SUCCESS in 8m 45s\n- openstack-tox-validate http://logs.openstack.org/28/666728/6/check/openstack-tox-validate/437d0f3/ : SUCCESS in 5m 40s\n- releases-tox-list-changes http://logs.openstack.org/28/666728/6/check/releases-tox-list-changes/4c9993b/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":6},{"id":"53e1586580f2d51e8c66b0eb28d717d9e687de41","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 11:34:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"99733db45282ef54067e058fd6e860a55cc1bfa6","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-11 12:07:23.000000000","message":"Patch Set 6: Code-Review+1\n\nVoting on my own patch, for the record of \"I voted for it!\"","accounts_in_message":[],"_revision_number":6},{"id":"5d230d7de7552295a6e04c1ead3ad1218a370712","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-11 16:38:55.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"887bdfb2c5f0343e7ddf3e89fbe35873877bfd4c","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-11 18:11:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"44d3b19a468798e8ea7372ed2ad63c9882ea58d1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-11 20:24:53.000000000","message":"Patch Set 6: Code-Review+1\n\nVoting on the same basis as evrardjp:)","accounts_in_message":[],"_revision_number":6},{"id":"9509e9a63ccce58d4bc115ea3dec919bf7b3e36b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 12:51:09.000000000","message":"Patch Set 6: Code-Review+1\n\nyay, ship it!","accounts_in_message":[],"_revision_number":6},{"id":"36359ad9c99f254e71ecf5b83ea7d6bf063aa7e9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-16 14:56:31.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8cdc767f188bd5e574981f80c88ca2928d3f17ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:56:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1b74840dbf9fdcd0857b7169c3d8fe4385ca5191","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:10:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/666728/6/gate/openstack-tox-docs/d252398/html/ : SUCCESS in 8m 09s\n- openstack-tox-validate http://logs.openstack.org/28/666728/6/gate/openstack-tox-validate/2a0d6e7/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":6},{"id":"82f7fdd628c40b97563f319c104efa200161bacd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:10:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"368dad9004b65621146a50ab98cb84cea6b3d7b8","revisions":{"55938a7424615f0842974db985730cb0ec7e9039":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:19:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/666728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/666728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/666728/1"}}},"commit":{"parents":[{"commit":"32bbe5121732e63b95ab96f68b3178c80ccc3d23","subject":"Release OpenStack-Ansible Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32bbe5121732e63b95ab96f68b3178c80ccc3d23"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-20 20:17:59.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-20 20:17:59.000000000","tz":180},"subject":"Release OpenStack-Ansible Stein","message":"Release OpenStack-Ansible Stein\n\nStein release based on the RC3\n\nChange-Id: I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/55938a7424615f0842974db985730cb0ec7e9039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/55938a7424615f0842974db985730cb0ec7e9039"}]},"branch":"refs/heads/master"},"9c295b939feb1ca0cc522f72bcb16f6d84ae1036":{"kind":"REWORK","_number":2,"created":"2019-06-21 05:16:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/666728/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/666728/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/666728/2"}}},"commit":{"parents":[{"commit":"32bbe5121732e63b95ab96f68b3178c80ccc3d23","subject":"Release OpenStack-Ansible Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/32bbe5121732e63b95ab96f68b3178c80ccc3d23"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-20 20:17:59.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-06-21 05:16:32.000000000","tz":0},"subject":"Release OpenStack-Ansible Stein","message":"Release OpenStack-Ansible Stein\n\nStein release based on the RC3\n\nChange-Id: I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c295b939feb1ca0cc522f72bcb16f6d84ae1036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c295b939feb1ca0cc522f72bcb16f6d84ae1036"}]},"branch":"refs/heads/master"},"1073335f0dba71d8b615d44eeeed9ce8323c5ddb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-27 13:21:26.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/28/666728/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/666728/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/666728/3"}}},"commit":{"parents":[{"commit":"4dd7ebd093b1038cefff78220f49b932f648ed96","subject":"Merge \"Release OpenStack-Ansible Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4dd7ebd093b1038cefff78220f49b932f648ed96"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-20 20:17:59.000000000","tz":180},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-06-27 13:21:26.000000000","tz":0},"subject":"Release OpenStack-Ansible Stein","message":"Release OpenStack-Ansible Stein\n\nStein release based on the RC3\n\nChange-Id: I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1073335f0dba71d8b615d44eeeed9ce8323c5ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1073335f0dba71d8b615d44eeeed9ce8323c5ddb"}]},"branch":"refs/heads/master"},"49ba84f6b02e52af7a42913da00d44f25e9c76fd":{"kind":"REWORK","_number":4,"created":"2019-07-10 14:45:39.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/28/666728/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/666728/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/666728/4"}}},"commit":{"parents":[{"commit":"4dd7ebd093b1038cefff78220f49b932f648ed96","subject":"Merge \"Release OpenStack-Ansible Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4dd7ebd093b1038cefff78220f49b932f648ed96"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-20 20:17:59.000000000","tz":180},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-10 14:45:37.000000000","tz":0},"subject":"Release OpenStack-Ansible Stein","message":"Release OpenStack-Ansible Stein\n\nStein release based on the RC3\n\nChange-Id: I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/49ba84f6b02e52af7a42913da00d44f25e9c76fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/49ba84f6b02e52af7a42913da00d44f25e9c76fd"}]},"branch":"refs/heads/master"},"272ea41f221caa1ca91bbda2a462af9c543efd1a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-10 14:46:23.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/28/666728/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/666728/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/666728/5"}}},"commit":{"parents":[{"commit":"4dd7ebd093b1038cefff78220f49b932f648ed96","subject":"Merge \"Release OpenStack-Ansible Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4dd7ebd093b1038cefff78220f49b932f648ed96"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-20 20:17:59.000000000","tz":180},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-10 14:46:20.000000000","tz":0},"subject":"Release OpenStack-Ansible Stein","message":"Release OpenStack-Ansible Stein\n\nStein release based on the head of the branch.\n\nChange-Id: I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/272ea41f221caa1ca91bbda2a462af9c543efd1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/272ea41f221caa1ca91bbda2a462af9c543efd1a"}]},"branch":"refs/heads/master"},"368dad9004b65621146a50ab98cb84cea6b3d7b8":{"kind":"REWORK","_number":6,"created":"2019-07-10 15:31:42.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/28/666728/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/666728/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/666728/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/666728/6"}}},"commit":{"parents":[{"commit":"9599bc26cd873e22e312b5d4d1a61ffff8183134","subject":"Release OpenStack-Ansible Stein RC4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9599bc26cd873e22e312b5d4d1a61ffff8183134"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-20 20:17:59.000000000","tz":180},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-10 15:30:58.000000000","tz":120},"subject":"Release OpenStack-Ansible Stein","message":"Release OpenStack-Ansible Stein\n\nStein release based on our final RC, RC4.\n\nCo-Authored-By: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nChange-Id: I3e82ee826ff7590a2f0757ddeceb4fc7886a5a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/368dad9004b65621146a50ab98cb84cea6b3d7b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/368dad9004b65621146a50ab98cb84cea6b3d7b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
