)]}'
{"id":"openstack%2Freleases~667111","triplet_id":"openstack%2Freleases~master~Icbe88daf7a6a86f16462b737cdb102fa517433df","project":"openstack/releases","branch":"master","topic":"os-traits-0.15.0","hashtags":[],"change_id":"Icbe88daf7a6a86f16462b737cdb102fa517433df","subject":"Release os-traits 0.15.0","status":"MERGED","created":"2019-06-24 14:18:41.000000000","updated":"2019-07-01 13:48:25.000000000","submitted":"2019-06-28 20:10:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667111-1561752640885-0a7135d0","meta_rev_id":"9ee96e1ea747a3ac67c8d3ae771be811f0eb58aa","_number":667111,"virtual_id_number":667111,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-06-27 14:21:48.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-06-28 20:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-28 13:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-06-28 19:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-28 19:56:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-28 13:34:27.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 09:10:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-28 13:34:27.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-28 19:53:59.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-28 19:56:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-28 20:10:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6b7c7f888822d9f730c9fe03c7ee205f64b9d66","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-24 14:18:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02fcd47c289ec972bc1c12eba5b20601941b3d90","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-24 14:19:32.000000000","message":"Patch Set 1:\n\nNote that PTL is on PTO this week.","accounts_in_message":[],"_revision_number":1},{"id":"e91abafdbc0625bda12fe1a14594bc3ee9e04470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:31:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/667111/1/check/openstack-tox-docs/ce6025d/html/ : SUCCESS in 7m 35s\n- openstack-tox-validate http://logs.openstack.org/11/667111/1/check/openstack-tox-validate/8712eeb/ : SUCCESS in 4m 11s\n- releases-tox-list-changes http://logs.openstack.org/11/667111/1/check/releases-tox-list-changes/708a55d/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a6e8e310a43ae1fc0edfbba89c88de0936ca012a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-27 09:10:54.000000000","message":"Patch Set 1: Code-Review+1\n\n@Eric, can we wait until his return and +1?","accounts_in_message":[],"_revision_number":1},{"id":"74396deb100c50675adb8dbd30c75ccacc5adeea","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 12:31:53.000000000","message":"Patch Set 1:\n\nYeah, not super urgent.","accounts_in_message":[],"_revision_number":1},{"id":"6f5b1b68144c04da56d9c702f5a8298a8fc61063","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 12:39:13.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s get COMPUTE_STATUS_DISABLED [1] into this release.\n\n[1] https://review.opendev.org/667209","accounts_in_message":[],"_revision_number":1},{"id":"8159d4d0c4e499333deb2547638ba5a56254b721","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 14:21:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e185d39d8750970d0429569c9e2948952b4e4eda","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 14:21:48.000000000","message":"Patch Set 2:\n\n\u003e Let\u0027s get COMPUTE_STATUS_DISABLED [1] into this release.\n\ndone","accounts_in_message":[],"_revision_number":2},{"id":"43afbed11dbb5e10547f68493cb8923b008b5e17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:59:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/667111/2/check/openstack-tox-docs/97d7eac/html/ : SUCCESS in 7m 22s\n- openstack-tox-validate http://logs.openstack.org/11/667111/2/check/openstack-tox-validate/c7d5697/ : SUCCESS in 4m 32s\n- releases-tox-list-changes http://logs.openstack.org/11/667111/2/check/releases-tox-list-changes/2f8b3a3/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e24b1be4740d7204bfe5123cb8b041be2d79808b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-28 13:34:41.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me, just need Chris\u0027s ack.","accounts_in_message":[],"_revision_number":2},{"id":"0c24203e0ca11f62caa6c26c0785fe76d19d8d0a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-28 19:53:59.000000000","message":"Patch Set 2: Code-Review+1\n\nI am a mighty placement core and will flex all over this thing.","accounts_in_message":[],"_revision_number":2},{"id":"dcd76980c70a13115bec4b0a2789b0a5a5f8786b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-28 19:56:15.000000000","message":"Patch Set 2: Workflow+1\n\nPer request from team, I\u0027m going to push this through on a Friday afternoon without the ack from the PTL to ease some pain.","accounts_in_message":[],"_revision_number":2},{"id":"7fce86068dea262cd48cbea940c3319e18ff8906","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:56:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af2ef7b4555bfab78f9471e609df7e371e8ea482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 20:10:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"05a67541454ee1cb9a6ec301d33e0085666a2877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 20:10:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/667111/2/gate/openstack-tox-docs/9de2baf/html/ : SUCCESS in 7m 25s\n- openstack-tox-validate http://logs.openstack.org/11/667111/2/gate/openstack-tox-validate/97a1fbf/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fd90a6fa85f17dde0d6f9e7152ccee1672a461ba","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 13:45:30.000000000","message":"Patch Set 2:\n\n\u003e Per request from team, I\u0027m going to push this through on a Friday\n \u003e afternoon without the ack from the PTL to ease some pain.\n\nAs a general rule, if Eric or Matt say go, you can take that as a proxy. We can make that official if people like.","accounts_in_message":[],"_revision_number":2},{"id":"cd3e7df9d61d7eb7450ba0640a87e4dd03ae319c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-01 13:48:25.000000000","message":"Patch Set 2:\n\n\u003e \u003e Per request from team, I\u0027m going to push this through on a Friday\n \u003e \u003e afternoon without the ack from the PTL to ease some pain.\n \u003e \n \u003e As a general rule, if Eric or Matt say go, you can take that as a\n \u003e proxy. We can make that official if people like.\n\nIf you want, and to avoid the implicit knowledge, you could add them both to https://opendev.org/openstack/releases/src/branch/master/data/release_liaisons.yaml#L114. That will make them show up in the list-changes report as official release liaisons and avoid any questions about who needs to give an ack to do a release.\n\nYou can also remove yourself from there if you do that since the PTL does not need to be listed twice.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f21aff290681adcd67e4f2c1814361d9b6e64b6","revisions":{"3c3e8b1a5530aa3af9737d734902d3b23b5dbe1a":{"kind":"REWORK","_number":1,"created":"2019-06-24 14:18:41.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/11/667111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/667111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/667111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/667111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/667111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/667111/1"}}},"commit":{"parents":[{"commit":"d6e7a480ea38a443492a95bf0d857c606a051abf","subject":"Merge \"Release python-ironic-inspector-client 3.6.0 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6e7a480ea38a443492a95bf0d857c606a051abf"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-24 14:17:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-24 14:17:15.000000000","tz":-300},"subject":"Release os-traits 0.15.0","message":"Release os-traits 0.15.0\n\nPicks up traits for image metadata [1] [2].\n\n[1] https://review.opendev.org/652996\n[2] https://review.opendev.org/665830\n\nChange-Id: Icbe88daf7a6a86f16462b737cdb102fa517433df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c3e8b1a5530aa3af9737d734902d3b23b5dbe1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c3e8b1a5530aa3af9737d734902d3b23b5dbe1a"}]},"branch":"refs/heads/master"},"6f21aff290681adcd67e4f2c1814361d9b6e64b6":{"kind":"REWORK","_number":2,"created":"2019-06-27 14:21:23.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/11/667111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/667111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/667111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/667111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/667111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/667111/2"}}},"commit":{"parents":[{"commit":"d6e7a480ea38a443492a95bf0d857c606a051abf","subject":"Merge \"Release python-ironic-inspector-client 3.6.0 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6e7a480ea38a443492a95bf0d857c606a051abf"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-24 14:17:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-27 14:19:55.000000000","tz":-300},"subject":"Release os-traits 0.15.0","message":"Release os-traits 0.15.0\n\nPicks up traits for image metadata [1] [2] and disabled computes [3].\n\n[1] https://review.opendev.org/652996\n[2] https://review.opendev.org/665830\n[3] https://review.opendev.org/667209\n\nChange-Id: Icbe88daf7a6a86f16462b737cdb102fa517433df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f21aff290681adcd67e4f2c1814361d9b6e64b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f21aff290681adcd67e4f2c1814361d9b6e64b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
