)]}'
{"id":"openstack%2Freleases~667685","triplet_id":"openstack%2Freleases~master~I3803a432da2538af4de163b77c192130366ddadf","project":"openstack/releases","branch":"master","topic":"ics","hashtags":[],"change_id":"I3803a432da2538af4de163b77c192130366ddadf","subject":"Fix iCalendar format violations","status":"MERGED","created":"2019-06-26 17:24:34.000000000","updated":"2019-06-27 09:18:55.000000000","submitted":"2019-06-27 09:18:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667685-1561627135375-31ea3c29","meta_rev_id":"5026943d84eae178ecfa4b6d5a4bdf3a4c7d52dc","_number":667685,"virtual_id_number":667685,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-26 19:01:34.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-06-27 09:18:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-27 04:28:46.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-06-27 09:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-27 09:09:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 04:28:46.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-27 09:09:17.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-27 09:18:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60c3fa77d215ab3bc805631462438afb54d97b76","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-26 17:24:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c8d4b112a1d9375797c699b2323d0b8cbe44794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/667685/1/check/openstack-tox-docs/15def45/html/ : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b342ee2853faf05e77a480409ea6249136e01c87","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-26 19:01:34.000000000","message":"Patch Set 1:\n\nFor convenience, here\u0027s the validation running on the results from this patch:\n\nhttps://icalendar.org/validator.html?url\u003dhttp://logs.openstack.org/85/667685/1/check/openstack-tox-docs/15def45/html/schedule.ics","accounts_in_message":[],"_revision_number":1},{"id":"1f243e2bfbefe2934ae60c1dc40f096590a2fdfe","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-27 04:28:46.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks Sean\n\nTIL: https://icalendar.org/validator.html ... I\u0027m scared to run it over the calendars on eavesdrop[1]\n\n\n[1] But I did[2]\n[2] https://icalendar.org/validator.html?url\u003dhttp://eavesdrop.openstack.org/irc-meetings.ical","accounts_in_message":[],"_revision_number":1},{"id":"4f4937b8ef82ba201fc9cedfcc65269b13407181","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-27 09:09:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nOuch, adding updating yaml2ical to my TODO","accounts_in_message":[],"_revision_number":1},{"id":"31e156c45e8fafe59309144ce470509b58a9593a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:09:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b23f3af50124f725ba903b48f2bf2ebb92c0eef4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:18:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"037efd697b9af454e41bb2f659681d19db748a1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:18:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/667685/1/gate/openstack-tox-docs/7b5a636/html/ : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1506d94374158344a543b208da39832d4166130f","revisions":{"1506d94374158344a543b208da39832d4166130f":{"kind":"REWORK","_number":1,"created":"2019-06-26 17:24:34.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/85/667685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/667685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/667685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/667685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/667685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/667685/1"}}},"commit":{"parents":[{"commit":"bde1ee920465db0cc2cbd5f4d10f03dc6d1e06c0","subject":"Set pike status to extended maintenance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bde1ee920465db0cc2cbd5f4d10f03dc6d1e06c0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-26 17:22:01.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-26 17:22:01.000000000","tz":-300},"subject":"Fix iCalendar format violations","message":"Fix iCalendar format violations\n\nAfter observing errors from importing our iCal into Outlook, I found\nthere were a few issues with our generated output according to [0].\n\nThis adds an overall VERSION to the iCal and sets per event timestamps\nand UIDs.\n\nThis also updates YAML loading to use safe_load to get rid of a\ndeprecation warning noticed while testing.\n\n[0] https://icalendar.org/validator.html?url\u003dhttps://releases.openstack.org/schedule.ics\n\nChange-Id: I3803a432da2538af4de163b77c192130366ddadf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1506d94374158344a543b208da39832d4166130f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1506d94374158344a543b208da39832d4166130f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
