)]}'
{"id":"openstack%2Freleases~668191","triplet_id":"openstack%2Freleases~master~Ic8711e68bea957e02be805523beac27daf3e70a8","project":"openstack/releases","branch":"master","topic":"tripleo/ansible/roles","hashtags":[],"change_id":"Ic8711e68bea957e02be805523beac27daf3e70a8","subject":"Updating tags for TripleO Ansible roles","status":"MERGED","created":"2019-06-28 15:22:36.000000000","updated":"2019-07-01 02:49:10.000000000","submitted":"2019-07-01 02:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668191-1561949350572-cfcade68","meta_rev_id":"70e57f593acdb9b7c8a6b9064dc0f3cbbb185154","_number":668191,"virtual_id_number":668191,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-01 02:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-28 17:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-01 02:39:33.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-06-28 15:25:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-01 02:39:33.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 15:25:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-28 17:32:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-01 02:39:33.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-01 02:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4db72b51abe5f9c5dabc1a1d57940c959f2ad5f7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-28 15:22:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07b8a2c43dd6e562d727acb8c772f1af7697dbf9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-28 15:25:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"39a339b2907532aee4fda05edd800acf3adf2572","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 15:35:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/668191/1/check/openstack-tox-docs/723ad13/html/ : SUCCESS in 6m 51s\n- openstack-tox-validate http://logs.openstack.org/91/668191/1/check/openstack-tox-validate/2e3e995/ : SUCCESS in 4m 39s\n- releases-tox-list-changes http://logs.openstack.org/91/668191/1/check/releases-tox-list-changes/8e47f2b/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"603754e2030e07f34b66a25d443057bcbfb3eacd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-28 17:32:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1bc41e962251c6dba5e6fc1e7343190fe5ddb42","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-01 02:39:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48656fa385a0894fce926b2a9fa4a3dd5b231678","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 02:39:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"335e04589af1813c8d68e6679e19c48135a106ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 02:49:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a1f07403bbe5ee7297972135c271294ea7c60ee0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 02:49:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/668191/1/gate/openstack-tox-docs/954409f/html/ : SUCCESS in 7m 12s\n- openstack-tox-validate http://logs.openstack.org/91/668191/1/gate/openstack-tox-validate/d3bdba6/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4be7fcc360c485a2a5776ee07e3eff6f4c1f5b02","revisions":{"4be7fcc360c485a2a5776ee07e3eff6f4c1f5b02":{"kind":"REWORK","_number":1,"created":"2019-06-28 15:22:36.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/91/668191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/668191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/668191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/668191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/668191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/668191/1"}}},"commit":{"parents":[{"commit":"7ebe023a141bde10714acd96bfccf72cb9b780cd","subject":"Merge \"training-labs stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ebe023a141bde10714acd96bfccf72cb9b780cd"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-28 15:19:25.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-28 15:19:25.000000000","tz":-240},"subject":"Updating tags for TripleO Ansible roles","message":"Updating tags for TripleO Ansible roles\n\n1. Cleanup\n\nansible-role-tripleo-modify-image and ansible-role-container-registry\nare indepenpent, we don\u0027t want branch them and only produce tags at this\npoint.\n\n2. New tags\n\nansible-role-chrony: Release 1.0.0, the role is now used by default in\nStein and proved to work fine.\n\nansible-role-container-registry: release 1.1.0\n\nansible-role-redhat-subscription: release 1.0.3\n\nChange-Id: Ic8711e68bea957e02be805523beac27daf3e70a8\nansible-role-tripleo-modify-image: release 1.1.0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4be7fcc360c485a2a5776ee07e3eff6f4c1f5b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4be7fcc360c485a2a5776ee07e3eff6f4c1f5b02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
