)]}'
{"id":"openstack%2Freleases~668471","triplet_id":"openstack%2Freleases~master~I697b956d11ce1463c2053b5f50fa625d4041cb44","project":"openstack/releases","branch":"master","topic":"rhsm","hashtags":[],"change_id":"I697b956d11ce1463c2053b5f50fa625d4041cb44","subject":"Release ansible-role-redhat-subscription","status":"MERGED","created":"2019-07-01 13:52:10.000000000","updated":"2019-07-04 10:10:04.000000000","submitted":"2019-07-04 10:10:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668471-1562235004362-6db76d6a","meta_rev_id":"c396dd9958aace0c0ee310dd22cd9e8846273a5c","_number":668471,"virtual_id_number":668471,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-04 10:10:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-02 17:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-04 09:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-01 14:03:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-04 09:49:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 14:03:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-02 17:55:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-04 09:49:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-04 10:10:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0799a47bf5d929260c1e8a8752d73c6a5d7061ea","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-01 13:52:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ceb956235d4edae4a96123c19d335948f5c484f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-01 14:03:48.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"8938245c6a3b601115f90d5cd2f59079dbc045d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:06:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/71/668471/1/check/openstack-tox-docs/12f7c4e/html/ : SUCCESS in 7m 47s\n- openstack-tox-validate http://logs.openstack.org/71/668471/1/check/openstack-tox-validate/a40ff35/ : SUCCESS in 4m 22s\n- releases-tox-list-changes http://logs.openstack.org/71/668471/1/check/releases-tox-list-changes/8aeed71/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f6bb3ec4689814d053b1a75d2d9a230d711acab0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-02 17:55:37.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks like that should fix it.","accounts_in_message":[],"_revision_number":1},{"id":"6143eccc9ce7a797d03b3005506b578a9d33d3fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-03 09:05:09.000000000","message":"Patch Set 1:\n\nFYI related to:\n- https://github.com/pypa/warehouse/issues/5890\n- https://github.com/pypa/warehouse/pull/5835\n\nThe same issue occured on pypi https://github.com/containers/python-podman/pull/37\n\nFYI the pypa team have reintroduce some checks on package deployment and especially long_description format type (markdown, restructuredText) and syntaxe checks to avoid to generate pypi project pages description in a raw format when something went wrong, now the pypa teéam reject deployments if something went wrong with the syntaxe parser to address it on the projects side and stop to publish documentation/descriptions in a bad format (like raw).\n\nWe need to specify the format that we use by using long_description_content_type \u003d text/markdown and my changes in fceb51c66eb343a7cd20891da8728c65ff703ce7 will introduce this config.\n\nAlso I\u0027ve submit few other patches on projects who can potentially facing the same issue https://review.opendev.org/#/q/status:open+topic:format-markdown\n\nI hope these informations can be useful for you.","accounts_in_message":[],"_revision_number":1},{"id":"f252bfb517cab5f5e35fa5f553e1228607e504c6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-04 09:49:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a28795f26235dfd838b7a0bbb850feac90d22aed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:49:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e844b302ec72df24ce156444e15c381c0b5f5222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:10:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"490ac044dfa5e7d155c344ebb2cea2f991b3227c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:10:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/71/668471/1/gate/openstack-tox-docs/6cab3ab/html/ : SUCCESS in 7m 30s\n- openstack-tox-validate http://logs.openstack.org/71/668471/1/gate/openstack-tox-validate/6710edf/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69599c77b372ecd93840a2667465a80c04c61905","revisions":{"69599c77b372ecd93840a2667465a80c04c61905":{"kind":"REWORK","_number":1,"created":"2019-07-01 13:52:10.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/71/668471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/668471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/668471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/668471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/668471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/668471/1"}}},"commit":{"parents":[{"commit":"2727ced0d5fbb69ea69472cbd7cded695bc5cff4","subject":"Merge \"Updating tags for TripleO Ansible roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2727ced0d5fbb69ea69472cbd7cded695bc5cff4"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-01 13:50:55.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-01 13:50:55.000000000","tz":-360},"subject":"Release ansible-role-redhat-subscription","message":"Release ansible-role-redhat-subscription\n\nThe last one failed during teh release process.\n\nChange-Id: I697b956d11ce1463c2053b5f50fa625d4041cb44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/69599c77b372ecd93840a2667465a80c04c61905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/69599c77b372ecd93840a2667465a80c04c61905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
