)]}'
{"id":"openstack%2Freleases~668864","triplet_id":"openstack%2Freleases~master~I7a2f5abc50361ef4690cf6eb0858dfc909eb45d8","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I7a2f5abc50361ef4690cf6eb0858dfc909eb45d8","subject":"Release oslotest 3.8.0","status":"MERGED","created":"2019-07-03 11:12:42.000000000","updated":"2019-07-05 12:41:28.000000000","submitted":"2019-07-05 12:41:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668864-1562330488090-a3f68f93","meta_rev_id":"37bb34fa0d21ce2467f923c3e84393a39b514775","_number":668864,"virtual_id_number":668864,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-07-05 12:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-07-03 18:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-07-05 12:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-04 09:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-07-05 12:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 18:56:38.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-04 09:52:36.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-05 12:26:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-05 12:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9f133023657f0a5e5510fdf94f7931935872bea","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 11:12:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"640a1c7d85d09b01d9ecb2a26852e83ebf520a90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 11:24:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/64/668864/1/check/openstack-tox-docs/ec8c527/ : POST_FAILURE in 5m 50s\n- openstack-tox-validate http://logs.openstack.org/64/668864/1/check/openstack-tox-validate/9d47394/ : FAILURE in 6m 35s\n- releases-tox-list-changes http://logs.openstack.org/64/668864/1/check/releases-tox-list-changes/9cf1db6/ : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f653a64379f5e3e3f059eeed7ee00f9015b16262","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-03 13:15:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\n+1 to releasing. I assume this is to unblock packaging since the mock change only affects the oslotest unit tests.","accounts_in_message":[],"_revision_number":1},{"id":"4734251be481a6b7d9e2be461323ff700d5ec945","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 15:17:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d73608806ba063c209fc5b42f5c0bde56c834e8","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 15:17:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3da85258899d6ba773ebb4d8c712a1e07f684799","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 15:17:32.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e +1 to releasing. I assume this is to unblock packaging since the\n \u003e mock change only affects the oslotest unit tests.\n\nyes. for packaging build failures.","accounts_in_message":[],"_revision_number":1},{"id":"1e68d387ff994243171b706ce90781e33dd2c9cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:29:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/64/668864/2/check/openstack-tox-docs/81efcd5/html/ : SUCCESS in 8m 04s\n- openstack-tox-validate http://logs.openstack.org/64/668864/2/check/openstack-tox-validate/b6f4825/ : SUCCESS in 4m 21s\n- releases-tox-list-changes http://logs.openstack.org/64/668864/2/check/releases-tox-list-changes/864238f/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"beb40f58bb5ff24f7410787653aa69306ccf1568","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-03 18:56:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9c78a4b16a2854a7cae8413f0ffd452ee5dfed45","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-04 09:52:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d85bd86599cecb6d88f034a79595a79b2f1a220a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-05 12:26:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3b420a298ee4f445f696a77888afc2fc9e142e63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 12:27:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eeb4f15ff4aefe640df35fc89a9db3b03ef0efe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 12:41:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/64/668864/2/gate/openstack-tox-docs/df51a10/html/ : SUCCESS in 7m 49s\n- openstack-tox-validate http://logs.openstack.org/64/668864/2/gate/openstack-tox-validate/56eb57c/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"37bb34fa0d21ce2467f923c3e84393a39b514775","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 12:41:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"707abef91b21133ca599bab9c95a302233ac5975","revisions":{"861bf618d1b9bba156e510ad95e966bdc0600608":{"kind":"REWORK","_number":1,"created":"2019-07-03 11:12:42.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/64/668864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/668864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/668864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/668864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/668864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/668864/1"}}},"commit":{"parents":[{"commit":"2727ced0d5fbb69ea69472cbd7cded695bc5cff4","subject":"Merge \"Updating tags for TripleO Ansible roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2727ced0d5fbb69ea69472cbd7cded695bc5cff4"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 11:11:38.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 11:11:38.000000000","tz":120},"subject":"Release oslotest 3.8.0","message":"Release oslotest 3.8.0\n\nThe release contains an important fix for newer mock versions.\n\nChange-Id: I7a2f5abc50361ef4690cf6eb0858dfc909eb45d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/861bf618d1b9bba156e510ad95e966bdc0600608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/861bf618d1b9bba156e510ad95e966bdc0600608"}]},"branch":"refs/heads/master"},"707abef91b21133ca599bab9c95a302233ac5975":{"kind":"REWORK","_number":2,"created":"2019-07-03 15:17:10.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/64/668864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/668864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/668864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/668864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/668864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/668864/2"}}},"commit":{"parents":[{"commit":"2727ced0d5fbb69ea69472cbd7cded695bc5cff4","subject":"Merge \"Updating tags for TripleO Ansible roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2727ced0d5fbb69ea69472cbd7cded695bc5cff4"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 11:11:38.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 15:17:03.000000000","tz":120},"subject":"Release oslotest 3.8.0","message":"Release oslotest 3.8.0\n\nThe release contains an important fix for newer mock versions.\n\nChange-Id: I7a2f5abc50361ef4690cf6eb0858dfc909eb45d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/707abef91b21133ca599bab9c95a302233ac5975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/707abef91b21133ca599bab9c95a302233ac5975"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
