)]}'
{"id":"openstack%2Freleases~669014","triplet_id":"openstack%2Freleases~master~Ife5d3a108a9ef566cbbd42a00b4859c240d77e64","project":"openstack/releases","branch":"master","topic":"nova-queens-17.0.11","hashtags":[],"change_id":"Ife5d3a108a9ef566cbbd42a00b4859c240d77e64","subject":"nova: release queens 17.0.11","status":"MERGED","created":"2019-07-04 01:32:13.000000000","updated":"2019-07-10 21:31:02.000000000","submitted":"2019-07-10 21:31:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669014-1562794262738-6e4dd5bd","meta_rev_id":"26308fb57ec5973b879f535790e61c6e7565fdc9","_number":669014,"virtual_id_number":669014,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2019-07-10 18:49:40.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-10 21:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-10 21:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2019-07-08 10:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-08 22:19:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-07-04 07:52:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-07-10 09:25:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-07-10 21:17:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-04 01:32:32.000000000","_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 01:32:32.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-07-04 07:52:23.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-07-08 10:22:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-08 22:19:01.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-10 09:25:14.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-10 21:17:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-10 21:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e48d47983c012a026a726b41de8f23d60708e0f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-04 01:32:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd7788fa2f24854b7bf3ad802f0d12de843543d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 01:53:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/669014/1/check/openstack-tox-docs/bf8a620/html/ : SUCCESS in 8m 08s\n- openstack-tox-validate http://logs.openstack.org/14/669014/1/check/openstack-tox-validate/daf087e/ : SUCCESS in 8m 49s\n- releases-tox-list-changes http://logs.openstack.org/14/669014/1/check/releases-tox-list-changes/0fe90d5/ : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d60088fc0675ae324c41fd526862b634c905481a","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-07-04 07:52:23.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, we did start blocking swap_volume with \u003e1 rw attachment as part of this but that\u0027s really just a bugfix so I\u0027m okay avoiding the minor bump for that. Everything else is pretty straight forward so +1 from me.","accounts_in_message":[],"_revision_number":1},{"id":"da99fdfeb5f300a7329514c23b83a792841ccbea","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-04 10:06:56.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, pausing to give Tony a chance to review for stable","accounts_in_message":[],"_revision_number":1},{"id":"0a749499ff9e96c4a4109ca96f6a648dd077de7d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-06 20:03:08.000000000","message":"Patch Set 1:\n\nLooks like Tony is on vacation for the next week, so we might need someone else to take a look. Doug? Sean?","accounts_in_message":[],"_revision_number":1},{"id":"fbf30fe2a39bcea59f74ec6fb42098eea5857a3f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-08 10:22:54.000000000","message":"Patch Set 1: Code-Review+2\n\nHad a shot at stable review and nothing stood out","accounts_in_message":[],"_revision_number":1},{"id":"7300ba2d5b89fc3681cd2cd9e741ed9c1fd70f82","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-08 22:19:01.000000000","message":"Patch Set 1: Code-Review+1\n\nThis lgtm, but it\u0027s one commit behind the latest.\n\n@Matt, any reason to bump this to include\n\n commit 95a35cbee44f8e97aa6d3a04820175c052f03cb3 (origin/stable/queens, gerrit/stable/queens)\n Merge: de7a8f9e44 9bc9567532\n Author: Zuul \u003czuul@review.opendev.org\u003e\n Date:   Thu Jul 4 17:10:51 2019 +0000\n\n    Merge \"Init HostState.failed_builds\" into stable/queens\n\n?","accounts_in_message":[],"_revision_number":1},{"id":"d640ca1c00c277fa98d64704eaf43f1537f9fea5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-09 19:56:23.000000000","message":"Patch Set 1:\n\n\u003e @Matt, any reason to bump this to include\n\nNo I specifically don\u0027t want to pick that up because it\u0027s not yet released on stein or rocky yet.","accounts_in_message":[],"_revision_number":1},{"id":"75d6ac060648cb8eefe0885f88ac61cee148bcfa","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 09:25:14.000000000","message":"Patch Set 1: Code-Review+1\n\nSensu stricto I am okay with the current version proposed. I was wondering for users if the version should be different due to the change of behaviour in \"Fail to live migration if instance has a NUMA topology\" but I think it\u0027s fine, as it was broken anyway. Need someone from stable to review this.","accounts_in_message":[],"_revision_number":1},{"id":"66a865ec80b5394cad85150d5dab6525d259bb30","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-10 18:49:40.000000000","message":"Patch Set 1:\n\n\u003e Need someone from stable to review this.\n\nDo I count? :)","accounts_in_message":[],"_revision_number":1},{"id":"8b18054cfe1064d7076b5abde8ad8184205d08be","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-10 21:17:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks fine to me. Since submitted by Matt, Tony is out, and it\u0027s been past our Monday stable review day, sending on through.","accounts_in_message":[],"_revision_number":1},{"id":"df0beaf49eb159d8c044b3718a29ba3bc940c9ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:17:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97bbd81ea6402d3b2c2cac060a540d3959296224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:31:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6a66688cf178184a2170402d79fa42e6ca37170b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:31:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/669014/1/gate/openstack-tox-docs/1c1cc0f/html/ : SUCCESS in 7m 30s\n- openstack-tox-validate http://logs.openstack.org/14/669014/1/gate/openstack-tox-validate/9bb582a/ : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dfe116dac5dc18bf2a31a51ccbbf1e6d7af276b","revisions":{"5dfe116dac5dc18bf2a31a51ccbbf1e6d7af276b":{"kind":"REWORK","_number":1,"created":"2019-07-04 01:32:13.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/14/669014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/669014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/669014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/669014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/669014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/669014/1"}}},"commit":{"parents":[{"commit":"1e2d8657c938b4dabf386f02175dd8a54ce48e91","subject":"Merge \"Release neutron-lib 1.28.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e2d8657c938b4dabf386f02175dd8a54ce48e91"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-07-04 01:31:10.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-07-04 01:31:59.000000000","tz":-240},"subject":"nova: release queens 17.0.11","message":"nova: release queens 17.0.11\n\nLots of bug fixes but nothing I\u0027d consider require\na minor version bump.\n\n$ git log --oneline --no-merges --no-decorate 17.0.10..\n9cbd2aedde fix up numa-topology live migration hypervisor check\n9999bce00f Fail to live migration if instance has a NUMA topology\nab445a027a Allow driver to properly unplug VIFs on destination on confirm resize\n8519b1ce6d Move get_pci_mapping_for_migration to MigrationContext\nd3bdeb2615 libvirt: Do not reraise DiskNotFound exceptions during resize\n2009859b52 Include all network devices in nova diagnostics\n20b90f2e26 Workaround missing RequestSpec.instance_group.uuid\n581df2c986 Add regression recreate test for bug 1830747\n95a3e0cf6a Restore connection_info after live migration rollback\n88fd40fe5d DRY up test_rollback_live_migration_set_migration_status\n58a140487c Block swap volume on volumes with \u003e1 rw attachment\n55deddf65d Fix live-migration when glance image deleted\n944b24f462 Add functional confirm_migration_error test\n5600309a1f [stable-only] Delete allocations even if _confirm_resize raises (part 2)\nb8435d6001 Delete allocations even if _confirm_resize raises\n0cad3089d3 Stop logging traceback when skipping quiesce\nf9c2503609 Fix retry of instance_update_and_get_original\n89063ecb61 Teardown networking when rolling back live migration even if shared disk\n01141392db Use migration_status during volume migrating and retyping\n99a63a6633 libvirt: Always disconnect volumes after libvirtError exceptions\n63b45a87b8 libvirt: Stop ignoring unknown libvirtError exceptions during volume attach\nba3ad2d65a Fix {min|max}_version in ironic Adapter setup\n22cd12dbb7 Create request spec, build request and mappings in one transaction\n5b0adaa0ca xenapi/agent: Change openssl error handling\nec43a1348d libvirt: Avoid using os-brick encryptors when device_path isn\u0027t provided\n899e881b18 Fix nova-grenade-live-migration run book for opendev migration\nd4dbbb645a Fix regression in glance client call\n1cee746085 OpenDev Migration Patch\n76418a32cb Move legacy-grenade-dsvm-neutron-multinode-live-migration in-tree\n5d5e7be8cb Fix legacy-grenade-dsvm-neutron-multinode-live-migration\nc8ad9f4e19 libvirt: set device address tag only if setting disk unit\nee52fb12bb libvirt: disconnect volume when encryption fails\n5433dc6a75 Add missing libvirt exception during device detach\nc6a42cd3b3 Error out migration when confirm_resize fails\nf1554d1b52 prevent common kwargs from glance client failure\ndba9b722af Define irrelevant-files for tempest-full-py3 job\n1b6f48354b Delete instance_id_mappings record in instance_destroy\n8f1773a7af Do not persist RequestSpec.ignore_hosts\n9fd4082d7c Add functional regression test for bug 1669054\n99cbbcfd61 Update instance.availability_zone on revertResize\n1b15a342e3 Add functional recreate test for bug 1819963\n50a0c6a4b9 [Stable Only] hardware: Handle races during pinning\nddc2c9118a Fix incomplete instance data returned after build failure\nee07c1c67d Update instance.availability_zone during live migration\n36d93675d9 Update resources once in update_available_resource\n705284e18c Don\u0027t persist zero allocation ratios in ResourceTracker\n7a3b072442 Document unset/reset wrinkle for *_allocation_ratio options\n1265e9e8e2 Replace openstack.org git:// URLs with https://\n4a7575bf14 Ignore VolumeAttachmentNotFound exception in compute.manager\n\nChange-Id: Ife5d3a108a9ef566cbbd42a00b4859c240d77e64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5dfe116dac5dc18bf2a31a51ccbbf1e6d7af276b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5dfe116dac5dc18bf2a31a51ccbbf1e6d7af276b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
