)]}'
{"id":"openstack%2Freleases~669185","triplet_id":"openstack%2Freleases~master~Id0946a6bbf86e79a4ba2a613c6950e3dd352cf80","project":"openstack/releases","branch":"master","topic":"list-deliverables-dates","hashtags":[],"change_id":"Id0946a6bbf86e79a4ba2a613c6950e3dd352cf80","subject":"Make display of tags optional","status":"MERGED","created":"2019-07-04 14:05:26.000000000","updated":"2019-07-19 05:03:40.000000000","submitted":"2019-07-19 05:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669185-1563512620474-5870c8eb","meta_rev_id":"1437c6f3f4e9302993465d9bf83d09312fb5cf89","_number":669185,"virtual_id_number":669185,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-19 05:03:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-05 12:28:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-19 04:50:57.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-07-10 08:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-19 04:50:57.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 12:28:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-10 08:45:02.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-19 04:50:57.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-19 05:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2b057405b66f17ba563aac920a56c6ff942766c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-04 14:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a53f9d4fbef06a13b1215d8105f76b02fab8c7a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:27:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/85/669185/1/check/openstack-tox-py35/b0ace13/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/85/669185/1/check/openstack-tox-py36/2d62b40/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/85/669185/1/check/openstack-tox-docs/a72582a/html/ : SUCCESS in 7m 04s\n- openstack-tox-validate http://logs.openstack.org/85/669185/1/check/openstack-tox-validate/f2e65fe/ : SUCCESS in 4m 34s\n- releases-tox-list-changes http://logs.openstack.org/85/669185/1/check/releases-tox-list-changes/f35f28b/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/85/669185/1/check/openstack-tox-pep8/0568763/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"28ae84cf94d9640258fd50136199a26acc38a10f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-05 12:28:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c67f35a68034d8f7dada208574c678f409ddb5b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 08:45:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c8cbb89e4097315ae8a745fd5894bc512d7f863","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 04:50:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbf0f475154320ab4d759e6a669083aff0ecc2fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 04:51:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67eb5e3ceb9934d4410c1a3f16aeb7e617295306","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 05:03:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d22562d3db01d3fa86bba177961dd7594e2f89f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 05:03:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/85/669185/1/gate/openstack-tox-py35/a4dae6e/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/85/669185/1/gate/openstack-tox-py36/8a61d6b/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/85/669185/1/gate/openstack-tox-docs/6cbff45/html/ : SUCCESS in 7m 45s\n- openstack-tox-validate http://logs.openstack.org/85/669185/1/gate/openstack-tox-validate/d88f291/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14d709176b6bae02e4bc613460e9ec18080c51ce","revisions":{"14d709176b6bae02e4bc613460e9ec18080c51ce":{"kind":"REWORK","_number":1,"created":"2019-07-04 14:05:26.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/85/669185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/669185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/669185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/669185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/669185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/669185/1"}}},"commit":{"parents":[{"commit":"2727ced0d5fbb69ea69472cbd7cded695bc5cff4","subject":"Merge \"Updating tags for TripleO Ansible roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2727ced0d5fbb69ea69472cbd7cded695bc5cff4"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-04 12:48:41.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-04 12:48:41.000000000","tz":120},"subject":"Make display of tags optional","message":"Make display of tags optional\n\nThe list of tags associated with a deliverable is rarely useful, and yet\ntakes up massive horizontal space in verbose mode. In preparation for\nadding more columns, make display of tags optional.\n\nChange-Id: Id0946a6bbf86e79a4ba2a613c6950e3dd352cf80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d709176b6bae02e4bc613460e9ec18080c51ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d709176b6bae02e4bc613460e9ec18080c51ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
