)]}'
{"id":"openstack%2Freleases~669186","triplet_id":"openstack%2Freleases~master~I00aff6703e85a00572edb363973c92b883a79456","project":"openstack/releases","branch":"master","topic":"list-deliverables-dates","hashtags":[],"change_id":"I00aff6703e85a00572edb363973c92b883a79456","subject":"List deliverables not released since given date","status":"MERGED","created":"2019-07-04 14:05:26.000000000","updated":"2019-07-19 05:21:47.000000000","submitted":"2019-07-19 05:21:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669186-1563513707294-6d7173bf","meta_rev_id":"66cdccb2290a89bf76f9ed4ace115c7dd7808006","_number":669186,"virtual_id_number":669186,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2019-07-11 08:47:31.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-07-19 05:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-12 20:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-19 05:02:35.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-14 13:17:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-07-13 13:46:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-07-14 13:24:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-19 05:02:35.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 08:43:34.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-12 20:46:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-13 13:46:19.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-14 13:17:37.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-14 13:24:03.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-19 05:02:35.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-19 05:21:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"beb633ce9060b6ec2a6c43670c6a660ee1154c17","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-04 14:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b44d4917455cb2fcc949147e8bbd6b4489a9c54f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:36:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/86/669186/1/check/openstack-tox-py35/4692829/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/86/669186/1/check/openstack-tox-py36/087b6e3/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/86/669186/1/check/openstack-tox-docs/d9933ac/html/ : SUCCESS in 7m 21s\n- openstack-tox-validate http://logs.openstack.org/86/669186/1/check/openstack-tox-validate/d298afc/ : SUCCESS in 3m 28s\n- releases-tox-list-changes http://logs.openstack.org/86/669186/1/check/releases-tox-list-changes/a493953/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/86/669186/1/check/openstack-tox-pep8/42f7454/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3932a57849a431019beea821c407682b04573457","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-05 12:23:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b0596cd68f30ca03488204db0a1f958f14583944","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 08:43:34.000000000","message":"Patch Set 1: Code-Review-1\n\nThis deserves a doc change in reference/using (helpers section). I think we should be more systematic there.","accounts_in_message":[],"_revision_number":1},{"id":"18e66a50d38c0b71db04345b3bb16921d9899b3f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 08:46:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16ee21d75b455d51808e9dbb711b5162516071eb","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 08:47:31.000000000","message":"Patch Set 2:\n\nSince this tool is meant to be used by the release team, I updated the process document rather than using document (which is targeted to PTLs and release liaisons)","accounts_in_message":[],"_revision_number":2},{"id":"74766a5f9e194243e4ca7c4c8ecae012844f177d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:15:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/86/669186/2/check/openstack-tox-py35/2129030/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/86/669186/2/check/openstack-tox-py36/f58a9f6/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/86/669186/2/check/openstack-tox-docs/363165f/html/ : SUCCESS in 7m 15s\n- openstack-tox-validate http://logs.openstack.org/86/669186/2/check/openstack-tox-validate/bdcd9db/ : SUCCESS in 5m 19s\n- releases-tox-list-changes http://logs.openstack.org/86/669186/2/check/releases-tox-list-changes/3574c78/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/86/669186/2/check/openstack-tox-pep8/95e2df6/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e6b5026b5bf010911b8d76487e814aba5415f3cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-12 20:46:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d4d720ae5b93787e3d00db6e0e91c727f8277c21","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-13 13:46:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7bb673a60fb4f194f6e02ff470364074637a1bfe","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 13:17:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bc8d84ee6311b16bb327e24a5a5cda949183ca76","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-14 13:24:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"74a14199576f7aa16c11076d317aa6db59955244","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 05:02:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63755071afd659861cb90fcd16defaf508398378","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 05:02:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c069c544e14dd7e71a2056c3ad5d01bee787d65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 05:21:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"45eb212ca052d1593cdf0faa54b9bad07e037442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 05:21:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/86/669186/2/gate/openstack-tox-py35/6077cd3/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/86/669186/2/gate/openstack-tox-py36/927d6ca/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/86/669186/2/gate/openstack-tox-docs/07a027f/html/ : SUCCESS in 7m 45s\n- openstack-tox-validate http://logs.openstack.org/86/669186/2/gate/openstack-tox-validate/0f061fb/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"845e10c0928c21c606ae15a754b5f93b886b5cb6","revisions":{"f989da3ca3f3ceb7a785b3912ae0ba72a1fb1f2f":{"kind":"REWORK","_number":1,"created":"2019-07-04 14:05:26.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/86/669186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/669186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/669186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/669186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/669186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/669186/1"}}},"commit":{"parents":[{"commit":"14d709176b6bae02e4bc613460e9ec18080c51ce","subject":"Make display of tags optional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d709176b6bae02e4bc613460e9ec18080c51ce"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-04 13:01:21.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-04 14:05:00.000000000","tz":120},"subject":"List deliverables not released since given date","message":"List deliverables not released since given date\n\nAllow list_deliverables to filter output to list only deliverables\nthat have not been released since a given date (--unreleased-since).\n\nThis will be useful to compile list of libraries to autorelease at\nmilestones, or check for cycle-with-intermediary services that may\nneed a refresh release.\n\nAdditionally, if --show-dates is provided, the dates for last releases\n(or all releases if -a option is provided) will be retrieved (and\ndisplayed if in verbose mode).\n\nRetrieving release dates is done through querying opendev\u0027s API for\ntag references and looking at committer dates, so it requires Internet\naccess and increases command run time. It is only enabled if\n--show_dates or --unreleased-since are enabled.\n\nTask: 35684\nChange-Id: I00aff6703e85a00572edb363973c92b883a79456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f989da3ca3f3ceb7a785b3912ae0ba72a1fb1f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f989da3ca3f3ceb7a785b3912ae0ba72a1fb1f2f"}]},"branch":"refs/heads/master"},"845e10c0928c21c606ae15a754b5f93b886b5cb6":{"kind":"REWORK","_number":2,"created":"2019-07-11 08:46:18.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/86/669186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/669186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/669186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/669186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/669186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/669186/2"}}},"commit":{"parents":[{"commit":"14d709176b6bae02e4bc613460e9ec18080c51ce","subject":"Make display of tags optional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d709176b6bae02e4bc613460e9ec18080c51ce"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-04 13:01:21.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-11 08:45:08.000000000","tz":120},"subject":"List deliverables not released since given date","message":"List deliverables not released since given date\n\nAllow list_deliverables to filter output to list only deliverables\nthat have not been released since a given date (--unreleased-since).\n\nThis will be useful to compile list of libraries to autorelease at\nmilestones, or check for cycle-with-intermediary services that may\nneed a refresh release.\n\nAdditionally, if --show-dates is provided, the dates for last releases\n(or all releases if -a option is provided) will be retrieved (and\ndisplayed if in verbose mode).\n\nRetrieving release dates is done through querying opendev\u0027s API for\ntag references and looking at committer dates, so it requires Internet\naccess and increases command run time. It is only enabled if\n--show_dates or --unreleased-since are enabled.\n\nTask: 35684\nChange-Id: I00aff6703e85a00572edb363973c92b883a79456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/845e10c0928c21c606ae15a754b5f93b886b5cb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/845e10c0928c21c606ae15a754b5f93b886b5cb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
